)]}'
{"id":"openstack%2Fpbr~660297","triplet_id":"openstack%2Fpbr~master~I486b6f645fa71f0fa519464465bf26411ca4faf4","project":"openstack/pbr","branch":"master","topic":"py37-job","hashtags":[],"change_id":"I486b6f645fa71f0fa519464465bf26411ca4faf4","subject":"Resolve some issue with tox.ini, setup.cfg","status":"MERGED","created":"2019-05-21 08:56:37.000000000","updated":"2019-07-15 16:00:09.000000000","submitted":"2019-07-15 16:00:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660297-1563206409026-4e3e88c1","meta_rev_id":"619fb8dc3c1dde4ec3760dc5e4e8613836f8c2c1","_number":660297,"virtual_id_number":660297,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-07-15 16:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-06-13 14:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-06-14 08:07:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-06-26 08:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-06-14 11:51:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-07-15 14:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2019-06-14 14:39:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-14 01:46:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":1,"date":"2019-07-15 14:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 14:59:48.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-14 01:46:11.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-14 08:07:16.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-06-14 11:51:17.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-06-14 14:39:43.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-26 08:55:06.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-07-15 14:07:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-15 16:00:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a830ace4238dd1e16368740d706e11f95eb7a87e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29999d05e3c5b33976dfc1bc956bb5407879c700","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:59:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/660297/1/check/test-release-openstack/3005fac/ : SUCCESS in 3m 57s\n- tempest-full http://logs.openstack.org/97/660297/1/check/tempest-full/5e1cff3/ : SUCCESS in 1h 50m 24s\n- tempest-full-py3 http://logs.openstack.org/97/660297/1/check/tempest-full-py3/0563bbf/ : SUCCESS in 1h 30m 02s\n- openstack-tox-cover http://logs.openstack.org/97/660297/1/check/openstack-tox-cover/321cebd/cover/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/1/check/openstack-tox-lower-constraints/a3574db/ : SUCCESS in 6m 17s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/1/check/openstack-tox-pep8/88de8d0/ : SUCCESS in 6m 46s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/1/check/openstack-tox-py27/107390c/ : SUCCESS in 7m 04s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/1/check/openstack-tox-py35/8cae505/ : SUCCESS in 7m 09s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/1/check/openstack-tox-py36/89875fe/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/1/check/openstack-tox-py37/984d58d/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/97/660297/1/check/openstack-tox-docs/63afc00/html/ : SUCCESS in 3m 27s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/1/check/pbr-installation-devstack/b0bc21b/ : FAILURE in 24m 52s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/1/check/pbr-installation-upstream-devstack/704a46d/ : FAILURE in 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5ca96ae42c2247d98909177f9dbb0a9ca1a01964","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 14:48:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cbc5ca81f1eb8f1231f679b6246d170f69f24dd0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 14:50:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c8705afe149f0ff0e56d87f2076fa45dd4b220a","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 14:59:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"866a51377fdf0650f60b4d570cafed1337e54e02","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-23 15:00:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7fafb96f356a26ec9cbdbff6f2eab518f34ce983","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-23 15:15:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f459299f31ae05575787ac0c87e8246da5a7861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 16:54:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/660297/2/check/test-release-openstack/af75183/ : SUCCESS in 3m 20s\n- tempest-full http://logs.openstack.org/97/660297/2/check/tempest-full/2017190/ : SUCCESS in 1h 52m 51s\n- tempest-full-py3 http://logs.openstack.org/97/660297/2/check/tempest-full-py3/bcfff47/ : SUCCESS in 1h 38m 45s\n- openstack-tox-cover http://logs.openstack.org/97/660297/2/check/openstack-tox-cover/bc1730b/cover/ : SUCCESS in 7m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/2/check/openstack-tox-lower-constraints/41919f7/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/2/check/openstack-tox-pep8/738ff12/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/2/check/openstack-tox-py27/b919c9e/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/2/check/openstack-tox-py35/99993cf/ : FAILURE in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/2/check/openstack-tox-py36/a01af59/ : SUCCESS in 9m 04s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/2/check/openstack-tox-py37/a328f18/ : SUCCESS in 4m 28s\n- openstack-tox-docs http://logs.openstack.org/97/660297/2/check/openstack-tox-docs/3c0b783/html/ : SUCCESS in 6m 18s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/2/check/pbr-installation-devstack/63fe4fc/ : SUCCESS in 30m 23s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/2/check/pbr-installation-upstream-devstack/10e17cd/ : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":2},{"id":"32a39cf6cff7f25a28b63a144a6e1d5dd87a9e7d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-24 13:52:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17b68cab0751d46c895ddc66700b5da9f950bfeb","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-24 16:30:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a298d568311baab05db03196631b900b97feae7f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-24 16:49:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74b2f2360df588ac56712cf2205b0bddf35d63c9","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-29 16:05:39.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"64f46861fa4d5f06d5f3088a57dd67b8e78dd8c4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-29 16:09:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df920ce0d3a768f48813738cb86a1c4c53f262e0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-29 16:12:18.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"98c87662c7bab10f230e20247dd56b0bc0a513b2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-29 16:12:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"29204c1d57eb81ead86e7df4da0121e70f24d05b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 17:59:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/660297/4/check/test-release-openstack/cccb2bd/ : SUCCESS in 3m 37s\n- tempest-full http://logs.openstack.org/97/660297/4/check/tempest-full/0c7622b/ : SUCCESS in 1h 34m 57s\n- tempest-full-py3 http://logs.openstack.org/97/660297/4/check/tempest-full-py3/3eb7262/ : SUCCESS in 1h 36m 03s\n- openstack-tox-cover http://logs.openstack.org/97/660297/4/check/openstack-tox-cover/16f3ca4/ : FAILURE in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/4/check/openstack-tox-lower-constraints/0a3b750/ : FAILURE in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/4/check/openstack-tox-pep8/effd40e/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/4/check/openstack-tox-py27/fc6bba4/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/4/check/openstack-tox-py35/b13d621/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/4/check/openstack-tox-py36/b1e0085/ : SUCCESS in 5m 17s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/4/check/openstack-tox-py37/5a7523f/ : SUCCESS in 4m 53s\n- openstack-tox-docs http://logs.openstack.org/97/660297/4/check/openstack-tox-docs/1f6279b/html/ : SUCCESS in 6m 19s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/4/check/pbr-installation-devstack/5f9ada0/ : SUCCESS in 30m 37s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/4/check/pbr-installation-upstream-devstack/35d9295/ : FAILURE in 14m 15s","accounts_in_message":[],"_revision_number":4},{"id":"7349bf3907378efaf91039fec95a3e0ff77515e8","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-05-29 18:05:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd4caa3946b46e1f09bf065a157bd9d508efd398","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-29 18:51:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cdad467958385964117077c2b63e098cc2462d18","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-30 10:14:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c99010a9349ca25b1500f8ef713ea3d17fa84346","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-30 10:18:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"993e3f5a67be724b8a246170090fe62eea96a94a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 12:27:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/660297/5/check/test-release-openstack/d2c4307/ : SUCCESS in 3m 40s\n- tempest-full http://logs.openstack.org/97/660297/5/check/tempest-full/568132d/ : SUCCESS in 1h 54m 15s\n- tempest-full-py3 http://logs.openstack.org/97/660297/5/check/tempest-full-py3/6daf7a0/ : SUCCESS in 1h 42m 37s\n- openstack-tox-cover http://logs.openstack.org/97/660297/5/check/openstack-tox-cover/348eb4d/cover/ : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/5/check/openstack-tox-lower-constraints/6087987/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/5/check/openstack-tox-pep8/6c502fb/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/5/check/openstack-tox-py27/69ae0a6/ : SUCCESS in 4m 55s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/5/check/openstack-tox-py35/e605f84/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/5/check/openstack-tox-py36/37b85ed/ : FAILURE in 8m 04s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/5/check/openstack-tox-py37/d8a4c5a/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/97/660297/5/check/openstack-tox-docs/943c912/html/ : SUCCESS in 3m 52s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/5/check/pbr-installation-devstack/4063660/ : SUCCESS in 25m 27s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/5/check/pbr-installation-upstream-devstack/1100759/ : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":5},{"id":"c031edc1efc27afc5e7d8327fcaa14778d75111b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-30 13:28:28.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"85244ec2db279a44e3ede9fedb6fd344c1608273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 15:37:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/660297/5/check/test-release-openstack/bde9506/ : SUCCESS in 3m 32s\n- tempest-full http://logs.openstack.org/97/660297/5/check/tempest-full/0fc48eb/ : SUCCESS in 1h 37m 58s\n- tempest-full-py3 http://logs.openstack.org/97/660297/5/check/tempest-full-py3/a7ffb07/ : SUCCESS in 1h 43m 39s\n- openstack-tox-cover http://logs.openstack.org/97/660297/5/check/openstack-tox-cover/81fe560/cover/ : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/5/check/openstack-tox-lower-constraints/0a1c0ab/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/5/check/openstack-tox-pep8/107a160/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/5/check/openstack-tox-py27/49cbf22/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/5/check/openstack-tox-py35/2d3850a/ : FAILURE in 7m 19s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/5/check/openstack-tox-py36/420ca34/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/5/check/openstack-tox-py37/72df13a/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/97/660297/5/check/openstack-tox-docs/66be4b6/html/ : SUCCESS in 4m 40s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/5/check/pbr-installation-devstack/c4fde38/ : SUCCESS in 27m 44s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/5/check/pbr-installation-upstream-devstack/8bb7a96/ : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":5},{"id":"388640b46bbf7e3eb0ef4597cf1046432ef4d17a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-13 14:46:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"580be36307770da0b164d3b4f1958381be371a05","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-13 14:59:48.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3aba867952ed366b3b0ed0e69604bed6a185efac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 18:34:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/660297/6/check/test-release-openstack/073f423/ : SUCCESS in 3m 24s\n- tempest-full http://logs.openstack.org/97/660297/6/check/tempest-full/b88658e/ : SUCCESS in 1h 42m 26s\n- tempest-full-py3 http://logs.openstack.org/97/660297/6/check/tempest-full-py3/dcd6493/ : TIMED_OUT in 2h 04m 37s\n- openstack-tox-cover http://logs.openstack.org/97/660297/6/check/openstack-tox-cover/839e1b5/cover/ : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/6/check/openstack-tox-lower-constraints/3594951/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/6/check/openstack-tox-pep8/6516493/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/6/check/openstack-tox-py27/18bc623/ : SUCCESS in 5m 25s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/6/check/openstack-tox-py35/7f094bb/ : SUCCESS in 6m 16s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/6/check/openstack-tox-py36/6f38237/ : SUCCESS in 7m 13s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/6/check/openstack-tox-py37/df18376/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/97/660297/6/check/openstack-tox-docs/af34260/html/ : SUCCESS in 5m 17s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/6/check/pbr-installation-devstack/0de1d7d/ : SUCCESS in 31m 01s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/6/check/pbr-installation-upstream-devstack/35feef8/ : FAILURE in 15m 59s","accounts_in_message":[],"_revision_number":6},{"id":"77ce9696b84133d8be242812d6f85b3b7cab5d6e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-13 18:59:23.000000000","message":"Patch Set 6:\n\nrecheck\n\ntimeout and an apt mirror issue","accounts_in_message":[],"_revision_number":6},{"id":"0f5055d3f3a5e6ee3ed661b63cc6f8914ac57cf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 20:54:31.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/97/660297/6/check/test-release-openstack/93ea1ef/ : SUCCESS in 4m 41s\n- tempest-full http://logs.openstack.org/97/660297/6/check/tempest-full/3c36bdc/ : FAILURE in 41m 03s\n- tempest-full-py3 http://logs.openstack.org/97/660297/6/check/tempest-full-py3/7eae568/ : FAILURE in 1h 43m 00s\n- openstack-tox-cover http://logs.openstack.org/97/660297/6/check/openstack-tox-cover/1fd98bc/cover/ : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/6/check/openstack-tox-lower-constraints/5350861/ : SUCCESS in 8m 54s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/6/check/openstack-tox-pep8/96d9d6a/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/6/check/openstack-tox-py27/227c801/ : SUCCESS in 5m 44s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/6/check/openstack-tox-py35/bc5ff00/ : SUCCESS in 11m 48s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/6/check/openstack-tox-py36/87e28ff/ : SUCCESS in 6m 07s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/6/check/openstack-tox-py37/4b46f32/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/97/660297/6/check/openstack-tox-docs/4b8574f/html/ : SUCCESS in 3m 56s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/6/check/pbr-installation-devstack/3de3441/ : SUCCESS in 34m 34s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/6/check/pbr-installation-upstream-devstack/46d27d1/ : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":6},{"id":"814befff5286556452c34069586139812a1fc2d5","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-14 01:46:11.000000000","message":"Patch Set 6: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"863f09faf5cf8f720fbd764d0d8d0d95c535e658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 03:37:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/97/660297/6/check/test-release-openstack/dcdcebd/ : SUCCESS in 3m 25s\n- tempest-full http://logs.openstack.org/97/660297/6/check/tempest-full/bbb5201/ : SUCCESS in 1h 43m 20s\n- tempest-full-py3 http://logs.openstack.org/97/660297/6/check/tempest-full-py3/8fe7e16/ : SUCCESS in 1h 45m 31s\n- openstack-tox-cover http://logs.openstack.org/97/660297/6/check/openstack-tox-cover/a2ffb84/cover/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/6/check/openstack-tox-lower-constraints/4d5826f/ : SUCCESS in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/6/check/openstack-tox-pep8/c95d563/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/6/check/openstack-tox-py27/e8904ff/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/6/check/openstack-tox-py35/7c28122/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/6/check/openstack-tox-py36/8689e4e/ : SUCCESS in 5m 44s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/6/check/openstack-tox-py37/1e3bbdc/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/97/660297/6/check/openstack-tox-docs/0eb5871/html/ : SUCCESS in 4m 09s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/6/check/pbr-installation-devstack/340d494/ : SUCCESS in 36m 39s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/6/check/pbr-installation-upstream-devstack/dcff9a1/ : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":6},{"id":"af492bdaad5fb47829cb53754f8ebcee9617c8b0","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-06-14 08:07:16.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2f2edde27668324d191b3d6b7ada0d45f2d6c42c","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-06-14 11:51:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1c26c474fb162adb2df232f64dc4094570452775","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:39:43.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"cb4d83ee72fa7f9569311afc11c2dbd213f347c1","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-06-26 08:55:06.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d08aa99c5ee160a688f3130ac207ee7511481251","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-15 14:07:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6534c6cdb7ee19dff2248b3be0d7afe34139195e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 14:08:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e2a73c5e3676571b00c6b3d9c5c8b698752c5f4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 16:00:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/97/660297/6/gate/test-release-openstack/8b1ce2e/ : SUCCESS in 2m 07s\n- tempest-full http://logs.openstack.org/97/660297/6/gate/tempest-full/2df588f/ : SUCCESS in 1h 45m 56s\n- tempest-full-py3 http://logs.openstack.org/97/660297/6/gate/tempest-full-py3/ea575ae/ : SUCCESS in 1h 28m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/97/660297/6/gate/openstack-tox-lower-constraints/e5e56ee/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/97/660297/6/gate/openstack-tox-pep8/6821784/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/97/660297/6/gate/openstack-tox-py27/7c0096e/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/97/660297/6/gate/openstack-tox-py35/d2a9e6d/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/97/660297/6/gate/openstack-tox-py36/19ff33f/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/97/660297/6/gate/openstack-tox-py37/e562806/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/97/660297/6/gate/openstack-tox-docs/f9c2dc8/html/ : SUCCESS in 3m 00s\n- pbr-installation-devstack http://logs.openstack.org/97/660297/6/gate/pbr-installation-devstack/f4da101/ : SUCCESS in 41m 06s\n- pbr-installation-upstream-devstack http://logs.openstack.org/97/660297/6/gate/pbr-installation-upstream-devstack/0699b42/ : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":6},{"id":"a679bc7cd5b8d271e3ee54338a7197ff64425982","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 16:00:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"79ef8d989a07e33303396a68b9d6fc19f22deeeb","revisions":{"62f9b0d403d33df0937e2e5cd1d5e41108dd8418":{"kind":"REWORK","_number":1,"created":"2019-05-21 08:56:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/660297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/97/660297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/97/660297/1"}}},"commit":{"parents":[{"commit":"4a5fbfa751d2cac786808341857c2d474d5cbc20","subject":"Add openstack-tox-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a5fbfa751d2cac786808341857c2d474d5cbc20"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:56:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:56:05.000000000","tz":60},"subject":"Resolve some issue with tox.ini, setup.cfg","message":"Resolve some issue with tox.ini, setup.cfg\n\n- Use constraints for documentation targets\n- Indicate support for Python 3.6, 3.7\n- Fix indentation\n\nChange-Id: I486b6f645fa71f0fa519464465bf26411ca4faf4\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/62f9b0d403d33df0937e2e5cd1d5e41108dd8418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/62f9b0d403d33df0937e2e5cd1d5e41108dd8418"}]},"branch":"refs/heads/master"},"3a670b5211295d36b2914722e664d2e9ad3e7114":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-23 14:48:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/660297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/97/660297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/97/660297/2"}}},"commit":{"parents":[{"commit":"20d95468bcc5df499b17e28b6ce440ad03a9a94f","subject":"Add openstack-tox-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/20d95468bcc5df499b17e28b6ce440ad03a9a94f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:56:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-23 14:48:58.000000000","tz":0},"subject":"Resolve some issue with tox.ini, setup.cfg","message":"Resolve some issue with tox.ini, setup.cfg\n\n- Use constraints for documentation targets\n- Indicate support for Python 3.6, 3.7\n- Fix indentation\n\nChange-Id: I486b6f645fa71f0fa519464465bf26411ca4faf4\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3a670b5211295d36b2914722e664d2e9ad3e7114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3a670b5211295d36b2914722e664d2e9ad3e7114"}]},"branch":"refs/heads/master"},"c44e8edbc8d68c3e85f9ea15a7e1e86392cecfaf":{"kind":"REWORK","_number":3,"created":"2019-05-29 16:09:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/660297/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/97/660297/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/97/660297/3"}}},"commit":{"parents":[{"commit":"20d95468bcc5df499b17e28b6ce440ad03a9a94f","subject":"Add openstack-tox-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/20d95468bcc5df499b17e28b6ce440ad03a9a94f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:56:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-29 16:09:20.000000000","tz":60},"subject":"Resolve some issue with tox.ini, setup.cfg","message":"Resolve some issue with tox.ini, setup.cfg\n\n- Use constraints for documentation targets\n- Indicate support for Python 3.6, 3.7\n- Fix indentation\n\nChange-Id: I486b6f645fa71f0fa519464465bf26411ca4faf4\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c44e8edbc8d68c3e85f9ea15a7e1e86392cecfaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c44e8edbc8d68c3e85f9ea15a7e1e86392cecfaf"}]},"branch":"refs/heads/master"},"ae81dfb93b1d90ba32ffdde9eb836fddbc88f20b":{"kind":"REWORK","_number":4,"created":"2019-05-29 16:12:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/660297/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/97/660297/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/97/660297/4"}}},"commit":{"parents":[{"commit":"20d95468bcc5df499b17e28b6ce440ad03a9a94f","subject":"Add openstack-tox-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/20d95468bcc5df499b17e28b6ce440ad03a9a94f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:56:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-29 16:10:30.000000000","tz":60},"subject":"Resolve some issue with tox.ini, setup.cfg","message":"Resolve some issue with tox.ini, setup.cfg\n\n- Use constraints for documentation targets\n- Indicate support for Python 3.6, 3.7\n- Fix indentation\n\nChange-Id: I486b6f645fa71f0fa519464465bf26411ca4faf4\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ae81dfb93b1d90ba32ffdde9eb836fddbc88f20b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ae81dfb93b1d90ba32ffdde9eb836fddbc88f20b"}]},"branch":"refs/heads/master"},"728bbb7159b41d97e66905d65ec40dafb2018e17":{"kind":"REWORK","_number":5,"created":"2019-05-30 10:18:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/660297/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/97/660297/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/97/660297/5"}}},"commit":{"parents":[{"commit":"20d95468bcc5df499b17e28b6ce440ad03a9a94f","subject":"Add openstack-tox-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/20d95468bcc5df499b17e28b6ce440ad03a9a94f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:56:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-30 10:18:49.000000000","tz":60},"subject":"Resolve some issue with tox.ini, setup.cfg","message":"Resolve some issue with tox.ini, setup.cfg\n\n- Use constraints for documentation targets\n- Indicate support for Python 3.6, 3.7\n- Fix indentation\n\nChange-Id: I486b6f645fa71f0fa519464465bf26411ca4faf4\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/728bbb7159b41d97e66905d65ec40dafb2018e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/728bbb7159b41d97e66905d65ec40dafb2018e17"}]},"branch":"refs/heads/master"},"79ef8d989a07e33303396a68b9d6fc19f22deeeb":{"kind":"REWORK","_number":6,"created":"2019-06-13 14:46:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/97/660297/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/97/660297/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/660297/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/97/660297/6"}}},"commit":{"parents":[{"commit":"17f9439e9f9026dd3e8cae1e917a78e80195152c","subject":"Merge \"Read description file as utf-8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/17f9439e9f9026dd3e8cae1e917a78e80195152c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-05-21 08:56:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-13 14:46:30.000000000","tz":60},"subject":"Resolve some issue with tox.ini, setup.cfg","message":"Resolve some issue with tox.ini, setup.cfg\n\n- Use constraints for documentation targets\n- Indicate support for Python 3.6, 3.7\n- Fix indentation\n\nChange-Id: I486b6f645fa71f0fa519464465bf26411ca4faf4\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/79ef8d989a07e33303396a68b9d6fc19f22deeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/79ef8d989a07e33303396a68b9d6fc19f22deeeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
