)]}'
{"id":"openstack%2Fpbr~663745","triplet_id":"openstack%2Fpbr~master~I8e3005e48247b9f277d9fd61a57adacaff1043d7","project":"openstack/pbr","branch":"master","topic":"wsgi-debug","hashtags":[],"change_id":"I8e3005e48247b9f277d9fd61a57adacaff1043d7","subject":"DNM: Debug wsgi ci failures","status":"ABANDONED","created":"2019-06-06 18:42:36.000000000","updated":"2019-06-10 16:05:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"24e514d062c2f239d5dd5b34218586b5577f0b0a","_number":663745,"virtual_id_number":663745,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-06-06 21:00:40.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":-1,"date":"2019-06-06 20:58:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 20:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16edaa49aea34f44b36743faba54b78f57fd3d64","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-06 18:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d27562bbdd79ad9a3e1eeb5932180f464ef2dc16","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-06 19:03:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"502d3962394127acdb9d1eb3df4b4b5840a7d5f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 19:30:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/45/663745/2/check/openstack-tox-cover/c08b865/cover/ : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/663745/2/check/openstack-tox-lower-constraints/68f9579/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/45/663745/2/check/openstack-tox-pep8/e9b9aa5/ : FAILURE in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/45/663745/2/check/openstack-tox-py27/07117cb/ : SUCCESS in 5m 24s\n- openstack-tox-py35 http://logs.openstack.org/45/663745/2/check/openstack-tox-py35/837b6ff/ : SUCCESS in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/45/663745/2/check/openstack-tox-py36/466450a/ : SUCCESS in 6m 51s\n- openstack-tox-py37 http://logs.openstack.org/45/663745/2/check/openstack-tox-py37/265cf03/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d3539364201376060306569453a4a48bae30ecb3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-06 19:41:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"269f6875ac7cc1bef6ddf75c71c233f962239416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 19:54:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/45/663745/2/check/openstack-tox-cover/db073f2/cover/ : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/663745/2/check/openstack-tox-lower-constraints/41a982d/ : SUCCESS in 5m 17s\n- openstack-tox-pep8 http://logs.openstack.org/45/663745/2/check/openstack-tox-pep8/1555ef6/ : FAILURE in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/45/663745/2/check/openstack-tox-py27/1860bce/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/45/663745/2/check/openstack-tox-py35/92b7f29/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/45/663745/2/check/openstack-tox-py36/d917ec0/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/45/663745/2/check/openstack-tox-py37/ff9224b/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b596ff91ee5d2ffc0e20d9f25f2a08b688bb4335","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-06 20:06:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0071bb1fa6433f7d7150a3d4a96d806cc9bcfda0","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-06 20:23:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cfbd04f00f914a2475372de20c51d7dc7dfe09b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 20:58:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/45/663745/4/check/openstack-tox-cover/fa04032/ : FAILURE in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/663745/4/check/openstack-tox-lower-constraints/4f0403e/ : FAILURE in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/45/663745/4/check/openstack-tox-pep8/c1880a8/ : FAILURE in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/45/663745/4/check/openstack-tox-py27/4b4cea7/ : FAILURE in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/45/663745/4/check/openstack-tox-py35/0fb3a2c/ : FAILURE in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/45/663745/4/check/openstack-tox-py36/185f711/ : FAILURE in 4m 51s\n- openstack-tox-py37 http://logs.openstack.org/45/663745/4/check/openstack-tox-py37/f794e5b/ : FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"a0e09a0d12f5e444f57a2ee4ba9c9182d6a5b140","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-06 21:00:40.000000000","message":"Patch Set 4:\n\nrecheck these all ran on rax nodes, would like to see results from some others.","accounts_in_message":[],"_revision_number":4},{"id":"99a15857747a745dd908c5ac45caed0d2b698469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:28:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/45/663745/4/check/openstack-tox-cover/aef10d8/ : FAILURE in 4m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/45/663745/4/check/openstack-tox-lower-constraints/fb9c6a9/ : FAILURE in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/45/663745/4/check/openstack-tox-pep8/59fdcb8/ : FAILURE in 6m 34s\n- openstack-tox-py27 http://logs.openstack.org/45/663745/4/check/openstack-tox-py27/b30bb47/ : FAILURE in 6m 11s\n- openstack-tox-py35 http://logs.openstack.org/45/663745/4/check/openstack-tox-py35/4f53192/ : FAILURE in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/45/663745/4/check/openstack-tox-py36/f76527e/ : FAILURE in 7m 37s\n- openstack-tox-py37 http://logs.openstack.org/45/663745/4/check/openstack-tox-py37/c1ecd48/ : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"b98cb4a5d5c998dd4fc466fb82afa06f3b73663e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-10 16:05:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"52f2bae2a2728e2594df068bff7012f575fbcb1b","revisions":{"dbdb2af729f0a803d748b0c54fdb9adcfc6e44ae":{"kind":"REWORK","_number":1,"created":"2019-06-06 18:42:36.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/45/663745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/45/663745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/45/663745/1"}}},"commit":{"parents":[{"commit":"c1e4225c5a5b9917c62f84d3ae56e747e2758f69","subject":"Merge \"Remove neutron-lbaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c1e4225c5a5b9917c62f84d3ae56e747e2758f69"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 18:42:15.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 18:42:15.000000000","tz":0},"subject":"DNM: Debug wsgi ci failures","message":"DNM: Debug wsgi ci failures\n\nChange-Id: I8e3005e48247b9f277d9fd61a57adacaff1043d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dbdb2af729f0a803d748b0c54fdb9adcfc6e44ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dbdb2af729f0a803d748b0c54fdb9adcfc6e44ae"}]},"branch":"refs/heads/master"},"9776213296d2c08515d0b9a26922bfaf112b8f6c":{"kind":"REWORK","_number":2,"created":"2019-06-06 19:03:12.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/45/663745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/45/663745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/45/663745/2"}}},"commit":{"parents":[{"commit":"c1e4225c5a5b9917c62f84d3ae56e747e2758f69","subject":"Merge \"Remove neutron-lbaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c1e4225c5a5b9917c62f84d3ae56e747e2758f69"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 18:42:15.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 19:03:05.000000000","tz":0},"subject":"DNM: Debug wsgi ci failures","message":"DNM: Debug wsgi ci failures\n\nChange-Id: I8e3005e48247b9f277d9fd61a57adacaff1043d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9776213296d2c08515d0b9a26922bfaf112b8f6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9776213296d2c08515d0b9a26922bfaf112b8f6c"}]},"branch":"refs/heads/master"},"e787478458f758d98c090aa723a2584bf541b342":{"kind":"REWORK","_number":3,"created":"2019-06-06 20:06:25.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/45/663745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/45/663745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/45/663745/3"}}},"commit":{"parents":[{"commit":"c1e4225c5a5b9917c62f84d3ae56e747e2758f69","subject":"Merge \"Remove neutron-lbaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c1e4225c5a5b9917c62f84d3ae56e747e2758f69"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 18:42:15.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 20:06:19.000000000","tz":0},"subject":"DNM: Debug wsgi ci failures","message":"DNM: Debug wsgi ci failures\n\nChange-Id: I8e3005e48247b9f277d9fd61a57adacaff1043d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e787478458f758d98c090aa723a2584bf541b342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e787478458f758d98c090aa723a2584bf541b342"}]},"branch":"refs/heads/master"},"52f2bae2a2728e2594df068bff7012f575fbcb1b":{"kind":"REWORK","_number":4,"created":"2019-06-06 20:23:14.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/45/663745/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/45/663745/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/45/663745/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/45/663745/4"}}},"commit":{"parents":[{"commit":"c1e4225c5a5b9917c62f84d3ae56e747e2758f69","subject":"Merge \"Remove neutron-lbaas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c1e4225c5a5b9917c62f84d3ae56e747e2758f69"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 18:42:15.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-06 20:21:54.000000000","tz":0},"subject":"DNM: Debug wsgi ci failures","message":"DNM: Debug wsgi ci failures\n\nChange-Id: I8e3005e48247b9f277d9fd61a57adacaff1043d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52f2bae2a2728e2594df068bff7012f575fbcb1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52f2bae2a2728e2594df068bff7012f575fbcb1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
