)]}'
{"id":"openstack%2Fpbr~663988","triplet_id":"openstack%2Fpbr~master~I5dc34e63f94ebf9c19ffb5caeeea24581eca658f","project":"openstack/pbr","branch":"master","topic":"wsgi-localhost","hashtags":[],"change_id":"I5dc34e63f94ebf9c19ffb5caeeea24581eca658f","subject":"Switch to release.o.o for constraints","status":"MERGED","created":"2019-06-07 16:05:30.000000000","updated":"2019-06-07 23:06:31.000000000","submitted":"2019-06-07 23:06:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663988-1559948791445-95369c51","meta_rev_id":"a2bb6655ce45298b6754b2623c54fb44c3ec3f5f","_number":663988,"virtual_id_number":663988,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-06-07 23:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-07 16:26:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-07 16:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-06-07 17:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 16:26:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-07 17:36:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-06-07 23:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f22f24cd9008f219d7bb0dd939aae0a183b6b26d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-07 16:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9620776421b7bd438c5a58010e01fe765f70696d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-07 16:06:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivial change. Fast approving","accounts_in_message":[],"_revision_number":1},{"id":"d0da975272c583e97dbb110597f0d6fe76d24df7","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-07 16:17:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fe9e8120f6f1e82220a32be477f142e12d787595","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-07 16:26:04.000000000","message":"Patch Set 2: Code-Review+1\n\nThis is the correct static URL for u-c.","accounts_in_message":[],"_revision_number":2},{"id":"8021e16ebe36133b6c987b1a3876841f8dae6abf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-07 17:36:06.000000000","message":"Patch Set 2: Workflow+1\n\nCounting Sean\u0027s +1 as a +2 since he knows things","accounts_in_message":[],"_revision_number":2},{"id":"15cca0e434f59a80ae50bdf64881a03839b103fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:40:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/88/663988/2/check/tempest-full/66e4ebf/ : SUCCESS in 1h 55m 48s\n- tempest-full-py3 http://logs.openstack.org/88/663988/2/check/tempest-full-py3/ecc7d76/ : SUCCESS in 1h 33m 37s\n- openstack-tox-cover http://logs.openstack.org/88/663988/2/check/openstack-tox-cover/423b7f3/cover/ : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/663988/2/check/openstack-tox-lower-constraints/bade65f/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/88/663988/2/check/openstack-tox-pep8/3665bd4/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/88/663988/2/check/openstack-tox-py27/4eda7e5/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/88/663988/2/check/openstack-tox-py35/20c505e/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/88/663988/2/check/openstack-tox-py36/e1f7105/ : SUCCESS in 5m 25s\n- openstack-tox-py37 http://logs.openstack.org/88/663988/2/check/openstack-tox-py37/d536a98/ : SUCCESS in 5m 21s\n- openstack-tox-docs http://logs.openstack.org/88/663988/2/check/openstack-tox-docs/bd1f3c7/html/ : SUCCESS in 4m 17s\n- pbr-installation-devstack http://logs.openstack.org/88/663988/2/check/pbr-installation-devstack/bb5a09c/ : SUCCESS in 27m 36s\n- pbr-installation-upstream-devstack http://logs.openstack.org/88/663988/2/check/pbr-installation-upstream-devstack/b6fc05a/ : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d21b29fc79af17b22f7b395c495831c9febcd510","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:40:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d49d655319ba243a7b6177244ccb6d6c7f63ab7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 21:07:27.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f57e2b56646bb592605e5729a4eed70e79ebca0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:06:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/88/663988/2/gate/tempest-full/b50a79c/ : SUCCESS in 1h 53m 02s\n- tempest-full-py3 http://logs.openstack.org/88/663988/2/gate/tempest-full-py3/48d2e69/ : SUCCESS in 1h 34m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/663988/2/gate/openstack-tox-lower-constraints/26f9fca/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/88/663988/2/gate/openstack-tox-pep8/0cf8f87/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/88/663988/2/gate/openstack-tox-py27/58a29c1/ : SUCCESS in 7m 25s\n- openstack-tox-py35 http://logs.openstack.org/88/663988/2/gate/openstack-tox-py35/b774b4c/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/88/663988/2/gate/openstack-tox-py36/3cebec3/ : SUCCESS in 6m 07s\n- openstack-tox-py37 http://logs.openstack.org/88/663988/2/gate/openstack-tox-py37/756a6a5/ : SUCCESS in 6m 54s\n- openstack-tox-docs http://logs.openstack.org/88/663988/2/gate/openstack-tox-docs/fed845c/html/ : SUCCESS in 4m 36s\n- pbr-installation-devstack http://logs.openstack.org/88/663988/2/gate/pbr-installation-devstack/b12fb47/ : SUCCESS in 35m 34s\n- pbr-installation-upstream-devstack http://logs.openstack.org/88/663988/2/gate/pbr-installation-upstream-devstack/9de8787/ : SUCCESS in 41m 59s","accounts_in_message":[],"_revision_number":2},{"id":"81ecffb7d4a3c95890c0ebcea0d62232c4db1d06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:06:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8e9da3d90d29aec71802f61c5270c3e5cf7cc90","revisions":{"d63ccc8b93e76e6b8e27c25f9b4d4ae659e82369":{"kind":"REWORK","_number":1,"created":"2019-06-07 16:05:30.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/88/663988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/663988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/663988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/663988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/663988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/663988/1"}}},"commit":{"parents":[{"commit":"49ad497ebe2de67ab4341e3f57ee7abb560e1945","subject":"Use sphinxcontrib-apidoc for api docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/49ad497ebe2de67ab4341e3f57ee7abb560e1945"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-07 16:05:04.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-07 16:05:04.000000000","tz":0},"subject":"Switch to release.o.o for constraints","message":"Switch to release.o.o for constraints\n\nThis is the preferred location to get constraints now.\n\nChange-Id: I5dc34e63f94ebf9c19ffb5caeeea24581eca658f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d63ccc8b93e76e6b8e27c25f9b4d4ae659e82369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d63ccc8b93e76e6b8e27c25f9b4d4ae659e82369"}]},"branch":"refs/heads/master"},"e8e9da3d90d29aec71802f61c5270c3e5cf7cc90":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-07 16:17:21.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/88/663988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/663988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/663988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/663988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/663988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/663988/2"}}},"commit":{"parents":[{"commit":"ab3db59c5aeac13f3998de09a2629e9fa52ec908","subject":"Make WSGI tests listen on localhost","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ab3db59c5aeac13f3998de09a2629e9fa52ec908"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-07 16:05:04.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-06-07 16:17:10.000000000","tz":0},"subject":"Switch to release.o.o for constraints","message":"Switch to release.o.o for constraints\n\nThis is the preferred location to get constraints now.\n\nChange-Id: I5dc34e63f94ebf9c19ffb5caeeea24581eca658f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e8e9da3d90d29aec71802f61c5270c3e5cf7cc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e8e9da3d90d29aec71802f61c5270c3e5cf7cc90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
