)]}'
{"id":"openstack%2Fpbr~669467","triplet_id":"openstack%2Fpbr~master~Ie9765e198e1204f56976476c8cfc9cc008c23d8f","project":"openstack/pbr","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ie9765e198e1204f56976476c8cfc9cc008c23d8f","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:14:38.000000000","updated":"2019-07-17 17:41:30.000000000","submitted":"2019-07-17 17:41:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669467-1563385290706-e47bf6dd","meta_rev_id":"93d4a0501b917fd7c64ab066d377f72efb295231","_number":669467,"virtual_id_number":669467,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-07-17 17:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-07-17 14:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-07-17 14:17:42.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-07-17 14:32:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 20:52:19.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-07-17 14:17:42.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-17 14:32:25.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-17 17:41:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50174cbb13da287007c6b0f46b81724b78bc5c06","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebb5c9c3e9787b037c3752fc0547de37247de6e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/67/669467/1/check/test-release-openstack/78ca89c/ : SUCCESS in 2m 04s\n- tempest-full http://logs.openstack.org/67/669467/1/check/tempest-full/4a6daa8/ : SUCCESS in 2h 03m 10s\n- tempest-full-py3 http://logs.openstack.org/67/669467/1/check/tempest-full-py3/143afd2/ : SUCCESS in 1h 35m 51s\n- openstack-tox-cover http://logs.openstack.org/67/669467/1/check/openstack-tox-cover/32ece14/cover/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/669467/1/check/openstack-tox-lower-constraints/34772c7/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/67/669467/1/check/openstack-tox-pep8/73fb322/ : SUCCESS in 2m 33s\n- openstack-tox-py27 http://logs.openstack.org/67/669467/1/check/openstack-tox-py27/5dff995/ : SUCCESS in 3m 40s\n- openstack-tox-py36 http://logs.openstack.org/67/669467/1/check/openstack-tox-py36/e57648d/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/67/669467/1/check/openstack-tox-py37/ff07c27/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/67/669467/1/check/openstack-tox-docs/97f2bd3/html/ : SUCCESS in 2m 28s\n- pbr-installation-devstack http://logs.openstack.org/67/669467/1/check/pbr-installation-devstack/49f2809/ : SUCCESS in 34m 26s\n- pbr-installation-upstream-devstack http://logs.openstack.org/67/669467/1/check/pbr-installation-upstream-devstack/c0fe260/ : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":1},{"id":"353da7619c316c815c46e78163974347e93a7245","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-08 15:28:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a281ee782ac41359d45f8ff92302a4a24ceeac72","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-08 16:14:43.000000000","message":"Patch Set 1:\n\nI\u0027d like to get Zane\u0027s input. I don\u0027t know how long py35 in the gate will be feasible.","accounts_in_message":[],"_revision_number":1},{"id":"8a0a96f70387bea8c93450a43e05c10acb4c9741","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:52:02.000000000","message":"Patch Set 1:\n\nAt some point having to support py35 causes problems for infra on stable branches. We\u0027re not at that point yet though, so AIUI we could maintain it for now with the caveat that we won\u0027t be able to support it until the end of stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"69c728193a44f6518e5751281526efa79b3b09fb","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-17 13:55:02.000000000","message":"Patch Set 1:\n\nKeeping py35 in the gate isn\u0027t a problem. Actually discussed with clarkb from infra yesterday wrt tempest keeping it and he said it\u0027s not a problem so let\u0027s keep it with pbr too. Thanks zane for your input.","accounts_in_message":[],"_revision_number":1},{"id":"349d65632f40e51de3885f0568380d0d0251cc92","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-17 14:03:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7495a4c6d5d4437bef522f4b936027900498e426","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-17 14:17:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b5b3c69ff1008674c8aa39419f1b2616704dfc52","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-17 14:32:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nCool, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"dd71ed6924a38fa4ccb3b90a34ac5243d9f43c16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 15:47:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/67/669467/2/check/tempest-full/5751ab9/ : SUCCESS in 1h 41m 49s\n- tempest-full-py3 http://logs.openstack.org/67/669467/2/check/tempest-full-py3/85f5b02/ : SUCCESS in 1h 16m 02s\n- openstack-tox-cover http://logs.openstack.org/67/669467/2/check/openstack-tox-cover/5d6816c/cover/ : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/669467/2/check/openstack-tox-lower-constraints/153a28d/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/67/669467/2/check/openstack-tox-pep8/842b3e1/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/67/669467/2/check/openstack-tox-py27/51f2092/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/67/669467/2/check/openstack-tox-py35/c381096/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/67/669467/2/check/openstack-tox-py36/af16a91/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/67/669467/2/check/openstack-tox-py37/951515d/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/67/669467/2/check/openstack-tox-docs/d27ccbd/html/ : SUCCESS in 3m 04s\n- pbr-installation-devstack http://logs.openstack.org/67/669467/2/check/pbr-installation-devstack/8e5aaf2/ : SUCCESS in 1h 05m 52s\n- pbr-installation-upstream-devstack http://logs.openstack.org/67/669467/2/check/pbr-installation-upstream-devstack/c79280a/ : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":2},{"id":"18f0af01ddb5db4b513b45c7ddc1265db9cbb8e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 15:47:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd306107fa8868065a55f5026bdcf7b1452ce447","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:41:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/67/669467/2/gate/tempest-full/490ce94/ : SUCCESS in 1h 46m 49s\n- tempest-full-py3 http://logs.openstack.org/67/669467/2/gate/tempest-full-py3/fbac7ae/ : SUCCESS in 1h 31m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/669467/2/gate/openstack-tox-lower-constraints/f52ca5b/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/67/669467/2/gate/openstack-tox-pep8/726bf8f/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/67/669467/2/gate/openstack-tox-py27/77ef079/ : SUCCESS in 4m 28s\n- openstack-tox-py35 http://logs.openstack.org/67/669467/2/gate/openstack-tox-py35/d519f3f/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/67/669467/2/gate/openstack-tox-py36/d7615df/ : SUCCESS in 4m 09s\n- openstack-tox-py37 http://logs.openstack.org/67/669467/2/gate/openstack-tox-py37/135713d/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/67/669467/2/gate/openstack-tox-docs/149af11/html/ : SUCCESS in 3m 15s\n- pbr-installation-devstack http://logs.openstack.org/67/669467/2/gate/pbr-installation-devstack/34a3133/ : SUCCESS in 32m 10s\n- pbr-installation-upstream-devstack http://logs.openstack.org/67/669467/2/gate/pbr-installation-upstream-devstack/3e1d346/ : SUCCESS in 29m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f833461aae3fd1b661950d9dd8777283abf56233","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 17:41:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4b89ae2cb3ea1af31c84c3df8419e3cad47d50c","revisions":{"f27d2a2927b2613fe650e9eea33e633ae3067559":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:14:38.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/67/669467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/67/669467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/67/669467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/67/669467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/67/669467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/67/669467/1"}}},"commit":{"parents":[{"commit":"c715db6df9d47b56cd31a3c65f934d1d220563bb","subject":"option to print only the version of a package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c715db6df9d47b56cd31a3c65f934d1d220563bb"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:59.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:14:37.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ie9765e198e1204f56976476c8cfc9cc008c23d8f\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f27d2a2927b2613fe650e9eea33e633ae3067559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f27d2a2927b2613fe650e9eea33e633ae3067559"}]},"branch":"refs/heads/master"},"e4b89ae2cb3ea1af31c84c3df8419e3cad47d50c":{"kind":"REWORK","_number":2,"created":"2019-07-17 14:03:17.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/67/669467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/67/669467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/67/669467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/67/669467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/67/669467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/67/669467/2"}}},"commit":{"parents":[{"commit":"aff2fd2d1d734f3a3b0851c81dd2b6c4cb8973bf","subject":"Merge \"Resolve some issue with tox.ini, setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/aff2fd2d1d734f3a3b0851c81dd2b6c4cb8973bf"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:59.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-17 14:03:07.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ie9765e198e1204f56976476c8cfc9cc008c23d8f\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e4b89ae2cb3ea1af31c84c3df8419e3cad47d50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e4b89ae2cb3ea1af31c84c3df8419e3cad47d50c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
