)]}'
{"id":"openstack%2Fpbr~678769","triplet_id":"openstack%2Fpbr~master~If0bc30e646a3bb35c02dc73cc31fb146b3ecd6b5","project":"openstack/pbr","branch":"master","topic":"python38","hashtags":[],"change_id":"If0bc30e646a3bb35c02dc73cc31fb146b3ecd6b5","subject":"Mark strings as raw","status":"MERGED","created":"2019-08-27 09:35:30.000000000","updated":"2019-11-13 16:20:45.000000000","submitted":"2019-11-13 16:18:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678769-1573661925785-bdb6f180","meta_rev_id":"9fad3264d7f16dee2849d287051774a7ccd417ec","_number":678769,"virtual_id_number":678769,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 16:18:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-08-27 19:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-09-27 11:33:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-11-13 15:09:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-11-13 14:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-11-13 14:44:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 19:31:17.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-09-27 11:33:53.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-11-13 14:44:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-13 15:09:21.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-13 16:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5ebcf15871254a67ce20a1491149fc7a355a5d4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-27 09:35:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9e27f520131e2ca1c0f771038d60c749ebd3e1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 11:30:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1daf188e74544ae492a8749d50b435cc : SUCCESS in 1h 49m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e290be737a64278a83df0fb06d425fb : SUCCESS in 1h 21m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89a5c1031f8d43a7a6417953fcf8db80 : FAILURE in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1ed0ab94100468e80ee56caf27221e5 : FAILURE in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd684a0505645f9b5d500bba7b33ba4 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf7b2198738444478aaa3420fdbb9c46 : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/41e123031de04824b58cd0e20721509d : FAILURE in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0734c720d67c418394803a4cf9d64f41 : FAILURE in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4df66b6b076b47378f91adf488669786 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d01aa1fdf4454ba68b2b98ca13391b34 : SUCCESS in 3m 48s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/8e400e3d32514b87a4d42d2ef838cbbe : SUCCESS in 48m 27s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/f8975bfd75c64987be23fa413e9a8240 : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":1},{"id":"015e06059a25de0ed337dc1c6cbcc6a51b823532","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-27 15:29:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nComment inline on the wsgi failure.","accounts_in_message":[],"_revision_number":1},{"id":"6f2588d94d83578ac15a5aba456e2de8aa07e17f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-27 15:54:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"601ab2931d439cd6d4eff94daaff95596c5cef7b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-27 15:57:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d5fa272f811b750aa0639b977673f7a0ad1e1a0","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-27 15:58:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdd127bbcb8e4f54f694fdb8d21d9a96714c93dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 17:42:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e4d5c3a26eff46d18ee13cecfe01617d : SUCCESS in 1h 43m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7228e07437d54f92a5663cc0e826c2b9 : SUCCESS in 1h 30m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f284f89691b94c0c9d612087105c1b53 : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c36919f3610d436fb04582c45b5aaf35 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f0445b1099488ebbbc2a384ae1169f : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b91d31dfb75b46ef8e90cc4c8f70237f : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/925e0406aa5c40769bbc3683b3aba520 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f1a386a41054ae89aec5a82548d9791 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f63830091aa24013aa41ab558fbe6696 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cd1e418da5741da83a128c746b03279 : SUCCESS in 4m 13s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/589f01f09dfd4111a27a947edc8e3d7a : SUCCESS in 41m 17s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/ddb7fdd4b3e742d2aa6f6c4eeee428ac : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ac6f6241b63bba382049e1a109a70fbf84608f9d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-27 19:31:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"19d35db28ccb1aaae2a12f5285bf87db29cd8ea0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-27 11:33:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"db1486383b6cadfdb800ec0ab93cd8c6aa9f99ad","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-13 14:44:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d676199c21a3613870eb6941827a08471351743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 14:44:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7204b64f5e197be38bea77601adb4ff13d72bd41","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 15:09:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ec7353f4b8a8628e2aadddfae0666b6055f5ef44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:18:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ab814ef1898249e3aeef03df18f99814 : SUCCESS in 1h 32m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8cad0a38adc246e9a030de8e43275a3e : SUCCESS in 1h 17m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0fbdddaf6444923909c03ef6c73c953 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cab85e9b9cb406aa11f569e26115b0e : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4e0df74391c417f99950acff2aab367 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/858422d7f27d4044ab2ce3ae74a5e768 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/218754a74df14b70a743f4137aa3a9e0 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa6ff3ad1475422daad0eb8445eb33e0 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35d142d1098405e8e02ea90dad53536 : SUCCESS in 4m 15s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/194716ca5fc24823ba9e81bdcd0643a9 : SUCCESS in 34m 26s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/b720cdb72cab4c56b796430162a9768a : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":2},{"id":"d1e0c81536db80ea0251488ba514661917ff3957","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:18:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7f7034b94d21c7f3053dd69501ddb2536a787d00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:20:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a115ce11f3a244a1ab1e4342ce18187f : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa0dd5652d82f8d238742f2d50c5514dec6f29a8","revisions":{"e360a17a6719aef3c0877879567f7e08b0650b90":{"kind":"REWORK","_number":1,"created":"2019-08-27 09:35:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/678769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/69/678769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/69/678769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/69/678769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/69/678769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/69/678769/1"}}},"commit":{"parents":[{"commit":"ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b","subject":"Merge \"Use more verbose logging for test failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-26 21:00:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-27 09:33:17.000000000","tz":60},"subject":"Mark strings as raw","message":"Mark strings as raw\n\nResolves the following deprecation warning:\n\n  DeprecationWarning: invalid escape sequence \\.\n\nChange-Id: If0bc30e646a3bb35c02dc73cc31fb146b3ecd6b5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e360a17a6719aef3c0877879567f7e08b0650b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e360a17a6719aef3c0877879567f7e08b0650b90"}]},"branch":"refs/heads/master"},"fa0dd5652d82f8d238742f2d50c5514dec6f29a8":{"kind":"REWORK","_number":2,"created":"2019-08-27 15:57:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/678769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/69/678769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/69/678769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/69/678769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/69/678769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/69/678769/2"}}},"commit":{"parents":[{"commit":"ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b","subject":"Merge \"Use more verbose logging for test failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-26 21:00:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-27 15:57:23.000000000","tz":60},"subject":"Mark strings as raw","message":"Mark strings as raw\n\nResolves the following deprecation warning:\n\n  DeprecationWarning: invalid escape sequence \\.\n\nChange-Id: If0bc30e646a3bb35c02dc73cc31fb146b3ecd6b5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fa0dd5652d82f8d238742f2d50c5514dec6f29a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fa0dd5652d82f8d238742f2d50c5514dec6f29a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
