)]}'
{"id":"openstack%2Fpbr~678770","triplet_id":"openstack%2Fpbr~master~If73238a39d0934c6f577fc52132fcbc63e8eb25f","project":"openstack/pbr","branch":"master","topic":"python38","hashtags":[],"change_id":"If73238a39d0934c6f577fc52132fcbc63e8eb25f","subject":"Update to latest hacking","status":"MERGED","created":"2019-08-27 09:35:30.000000000","updated":"2019-11-13 16:22:58.000000000","submitted":"2019-11-13 16:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678770-1573661926651-ebba688c","meta_rev_id":"5fd940254d12db14a6ae22b3271573a605fc53b0","_number":678770,"virtual_id_number":678770,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2019-08-28 08:33:44.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 16:18:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-08-27 19:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-11-13 15:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-11-13 14:48:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-11-13 14:48:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 19:35:31.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-11-13 14:48:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-13 15:07:26.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-13 16:18:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98f40ed979eefc24d554ec8f82d7c903ea63b5c4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-27 09:35:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57dceb76fbfd8fd82592a7ac1e831af41e4849f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 11:38:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e9575a7acb240c2b24079ae1b0ff59e : SUCCESS in 1h 33m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2332ff3fe5b54a428a8aa7b715816c71 : SUCCESS in 1h 56m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c84b1c9cf1a8458c90d088fe312c94cb : FAILURE in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a9f66d25035487391a152fdcf1fc97e : FAILURE in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/989746d9908f4c17b50e1f2c58c29720 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed50c691ec554f5497f4a425d60b42b0 : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a2e6cf4b693b42b4a335780fdc4daea1 : FAILURE in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03732a97855d443688e7fe6866f4de27 : FAILURE in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eecdce1339254303a8107b03995e6fd3 : FAILURE in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c1bd5e51984f719fb46d9a4bb9e1db : SUCCESS in 4m 19s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/bd9b5d95fad34847a9e55813637181d2 : SUCCESS in 48m 02s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/1a1dee3a3672444bb204136929fb525b : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b400b518a290edd217e6c3e2e82b337f4bda3a60","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-27 15:57:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6ec48c9d5d5863a194dc193993fb634555977998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:23:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9050c96e535840489f4b7544340010eb : SUCCESS in 1h 39m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef999a9fe0e748098e11e2f2b5cb493b : SUCCESS in 1h 45m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/433c1bfba67e4c08895c9622b59c666e : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4739be1d44834bdabab687e48bb877c9 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5598d6699a134290a1956879825e1e83 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a386b34165df4b078111fcde2a8b9cd6 : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2d82c952e07d41f3a01e178d0a9f8e82 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cecda561fff64d83ae79833fda2a829d : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/917c181f741546379b35ce70252d7f89 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fc0bf54c2844c6eb9a17cc5d05373ae : SUCCESS in 5m 44s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/23f6003b7a4241c888bd22ec2aeb0581 : SUCCESS in 44m 24s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/17d4d8c2b8434ba898fcce27f5863e3f : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":2},{"id":"de3ca1f50185347cad744c91c5576d361357c602","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-27 19:35:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aca5b8a63b43957233eca1f15530271dc60628c6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-28 08:33:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e31d4a788f3d6633303f468c388d3a15de776d8","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-28 13:33:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b74463c52b06d985950b35df67d3e6d5a3aac1b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-13 14:48:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b89442a78c1bef69ac8391c921c3249c51c4cd3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 14:48:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"980145cd82377e230b8b87ac695730ba9243643a","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 15:07:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4c5108b6f20810fa6a46025f8ad91b33fc6aa06d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:18:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2e4a2fa43fd5454187372cf66e74239b : SUCCESS in 1h 25m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a59d6395b8ff4a98afab6768573010cc : SUCCESS in 1h 17m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2635101415c94cccb51dc99d76effccf : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7067132ef3284e018844cedd7047f6c5 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/037197ee5fba4fe386541932a937fd6e : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b52cf9998d7b466c907ab9e151600ad5 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/908663f570714f29beb5102e7de3e71b : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed32204efbfc4bd28255ebabb283b6bc : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9267f0376c4539b1048b7137e0dbb0 : SUCCESS in 4m 32s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/36070fb9b51f4daa8194704e8b73a108 : SUCCESS in 41m 07s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/61f311404d794d9b94aabcb16105d335 : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b30c2a611eb30e268a54aad456d177048f75d446","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:18:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"28e15ba963d26f5339fb63c10f0938f997bb7c40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 16:22:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b1cf5eb36542a0afd4454f5a8948ba : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57d7c53739487e9fa0051ae12d8609acebaaf0e1","revisions":{"fef3aa615f4facf26ab59df7e53fcf56d3c36b95":{"kind":"REWORK","_number":1,"created":"2019-08-27 09:35:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/678770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/70/678770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/678770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/678770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/678770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/70/678770/1"}}},"commit":{"parents":[{"commit":"e360a17a6719aef3c0877879567f7e08b0650b90","subject":"Mark strings as raw","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e360a17a6719aef3c0877879567f7e08b0650b90"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-26 21:00:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-27 09:34:50.000000000","tz":60},"subject":"Update to latest hacking","message":"Update to latest hacking\n\nChange-Id: If73238a39d0934c6f577fc52132fcbc63e8eb25f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fef3aa615f4facf26ab59df7e53fcf56d3c36b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fef3aa615f4facf26ab59df7e53fcf56d3c36b95"}]},"branch":"refs/heads/master"},"57d7c53739487e9fa0051ae12d8609acebaaf0e1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-27 15:57:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/678770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/70/678770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/678770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/678770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/678770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/70/678770/2"}}},"commit":{"parents":[{"commit":"fa0dd5652d82f8d238742f2d50c5514dec6f29a8","subject":"Mark strings as raw","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fa0dd5652d82f8d238742f2d50c5514dec6f29a8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-26 21:00:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-08-27 15:57:23.000000000","tz":60},"subject":"Update to latest hacking","message":"Update to latest hacking\n\nChange-Id: If73238a39d0934c6f577fc52132fcbc63e8eb25f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/57d7c53739487e9fa0051ae12d8609acebaaf0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/57d7c53739487e9fa0051ae12d8609acebaaf0e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
