)]}'
{"id":"openstack%2Fpbr~691456","triplet_id":"openstack%2Fpbr~master~I059ae377aed10e5dcb0e418c6aa427aca88cf9e1","project":"openstack/pbr","branch":"master","hashtags":[],"change_id":"I059ae377aed10e5dcb0e418c6aa427aca88cf9e1","subject":"Switch to Ussuri jobs","status":"MERGED","created":"2019-10-26 03:22:05.000000000","updated":"2019-12-06 16:49:27.000000000","submitted":"2019-12-06 16:44:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691456-1575650656268-758459c9","meta_rev_id":"3adde180fe962c47416435821924a4cdd81c513e","_number":691456,"virtual_id_number":691456,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 16:44:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-12-06 15:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-12-03 08:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-12-06 15:11:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 08:32:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-06 15:11:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-06 16:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38d47fdb2b2aafd950102b079d98393ebe037bf2","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-26 03:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"147ec5a08edf81ec88cb6927de2f48307e20bb1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-26 05:02:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bea0f9b09a134b40aa93584265c1cf01 : SUCCESS in 1h 39m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffa3ac8c28f84a85b6995ef27768bcde : SUCCESS in 1h 14m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ae41789c8ed4a9aadd86e39b2911a63 : SUCCESS in 3m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6228a9af0a36407e894b531095af20d3 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c73908124ed42c99eefc59797e946e8 : SUCCESS in 2m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14462dc3755649dda41312bba557bf24 : SUCCESS in 3m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/535e3d0152b74a8b82c3f63283126ddf : SUCCESS in 2m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2ad0f7659249558cc4f5c45e9a077a : FAILURE in 8m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9da39ea260f54e41a3bad3b9e433d4ce : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bba312c150249f59e7bc33d747c0c20 : SUCCESS in 5m 15s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/d7b6243348ae40698eee37f20dc877e1 : SUCCESS in 27m 47s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/2d4bdd03b7464b42a95591a8f8f4bb92 : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"416783536a593e0930134f6ac76d4a5e6478ac7e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-02 18:16:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a473634206da0866e69b712925b6e1b270f9f045","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-12-03 02:47:49.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\nOK， I drop it, Thanks very much ,please check again","accounts_in_message":[],"_revision_number":1},{"id":"084b06296424473cb66ce7c84b6515a193dd7c18","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-12-03 02:49:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e8f512c063e29c23d11b477614f0e6062e29e27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 04:31:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/593e475b1f9f43ababfb0929845dff21 : SUCCESS in 1h 40m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a39cd176d2c4ea98d94e38eaf054ed0 : SUCCESS in 1h 18m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa1925522b9494dbc6d0e3238074aab : SUCCESS in 3m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a9deffbc4834ab0b1045559ba977cf1 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd897c45f4ed435b9b653f02af9f4f89 : SUCCESS in 2m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e08924d942ce455388206402199583f4 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64c74e7bf3aa4ce19a05e85169247c76 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/180c5f596d6b46558781b174cd4e9fb7 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6809ef83438a4a09a2b5b82201a11ec7 : SUCCESS in 2m 41s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/63c80d2e21894f948ef89d4a923dd675 : SUCCESS in 31m 08s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/3189ee08ff4f4ec28a7ca658f2e50cb3 : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":2},{"id":"db8e354e04fc4c6abff32391dd3449dfddb10b52","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-03 08:32:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba2f3dfd033d1f4a16243d177b63619975f7ce27","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-06 15:11:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb2970292e0c1663f38d8a6b11da3e4d48758f86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 15:11:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6bdd0d7ccfdf4ba50cf786eeae838a644734ad8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 16:44:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"679a2951fc3b33bfef70f74340b2fbdcbe2e9d44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 16:44:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/423b41e0002244889478379e76aef1ea : SUCCESS in 1h 27m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1262f3bc7b8147c4afec6d36c9696ae2 : SUCCESS in 1h 26m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd729bcea92a4c1f831a129e616e2991 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a82ee089c5497c9ef2e45c46ba05f5 : SUCCESS in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98a3d81eb5ae4fa1b1e11534f0d89467 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/406ceacaebb844c7a7c534b94f6c86b7 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c6d252bcd2c4bb6a540237ffb130be5 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca66d16f93a4e6b9fbf2ce1055e1bdb : SUCCESS in 4m 51s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/bc8abd1570b34350af69f3ac5f37757a : SUCCESS in 36m 29s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/fe629f52676d4706bc1bf17e18e2a531 : SUCCESS in 26m 25s","accounts_in_message":[],"_revision_number":2},{"id":"3adde180fe962c47416435821924a4cdd81c513e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 16:49:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a512b439fc8044368db8cd703bc6aa6f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b821df5f232e0843aced66b508c1d7355887bc9b","revisions":{"5835fbad28f7155a5e97a19608d82fcfd7b693a1":{"kind":"REWORK","_number":1,"created":"2019-10-26 03:22:05.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/56/691456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/56/691456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/691456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/691456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/691456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/56/691456/1"}}},"commit":{"parents":[{"commit":"ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b","subject":"Merge \"Use more verbose logging for test failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-26 03:21:58.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-26 03:21:58.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I059ae377aed10e5dcb0e418c6aa427aca88cf9e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5835fbad28f7155a5e97a19608d82fcfd7b693a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5835fbad28f7155a5e97a19608d82fcfd7b693a1"}]},"branch":"refs/heads/master"},"b821df5f232e0843aced66b508c1d7355887bc9b":{"kind":"REWORK","_number":2,"created":"2019-12-03 02:49:02.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/56/691456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/56/691456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/691456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/691456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/691456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/56/691456/2"}}},"commit":{"parents":[{"commit":"ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b","subject":"Merge \"Use more verbose logging for test failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ef2269d5dbe3b9f6b690f51f94f8c09f23bdb11b"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-26 03:21:58.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-12-03 02:46:48.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I059ae377aed10e5dcb0e418c6aa427aca88cf9e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b821df5f232e0843aced66b508c1d7355887bc9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b821df5f232e0843aced66b508c1d7355887bc9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
