)]}'
{"id":"openstack%2Fpbr~742160","triplet_id":"openstack%2Fpbr~master~I730018e2bf788be5ac4e993af1746296206edfc1","project":"openstack/pbr","branch":"master","topic":"oslo-pre-commit","hashtags":[],"change_id":"I730018e2bf788be5ac4e993af1746296206edfc1","subject":"Adding pre-commit","status":"MERGED","created":"2020-07-21 11:44:33.000000000","updated":"2020-10-22 00:26:37.000000000","submitted":"2020-10-22 00:24:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742160-1603326269899-dafe9f05","meta_rev_id":"b20a178c164ec6f5d90a8bdf5797a95f3b1d2c10","_number":742160,"virtual_id_number":742160,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 00:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-21 22:33:58.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-10-21 21:32:54.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":-1,"date":"2020-10-21 16:48:08.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-10-21 22:33:58.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 15:32:10.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-10 23:31:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-10-21 21:32:54.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2020-10-21 22:33:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-22 00:24:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d0c5cb704d58637b4514aae6726550c3c51374f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-21 11:44:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51870dfce05e3cf1edb135a0875cae0d2765f99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 13:13:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7cd5530b4d5d4fc9994690ec12a8a6f8 : SUCCESS in 1h 22m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93ecb8b0a7994f8fa8536337a0fe8278 : SUCCESS in 1h 22m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c518c7c4c0684802af89d14ae8f423e4 : FAILURE in 7m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/260c77ad8b654e1389791c1473408d4d : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebde24df87074d16b8ce6994f38b255f : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45b908f05fbe48bba681c73709c6e72c : FAILURE in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a69adedae41447e7a4b2445b336436c4 : FAILURE in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a9e2ba0d38949e09761497145f18d8d : FAILURE in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62608223ae254f8baf97ae60fae4b7c7 : FAILURE in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6301cee23e9d4bc2bcb6d753598b0586 : SUCCESS in 3m 36s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/4f29ae29fe154cb48ec1a863322c7595 : FAILURE in 22m 51s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/8c0399b8468a48afab2fa847d73e7e3e : FAILURE in 26m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9e4651f7a965841dda7a87611076147fed395d1c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-21 14:33:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36db3b3699e54277aa821dd35d59b4be32a5a4c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 16:08:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ce34b517ea1c4b3ab121e521447054bd : SUCCESS in 1h 30m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afcf1fdfcae84a81a5dca56f9cfb70df : SUCCESS in 1h 19m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daa93c7b4bf7467ea2228b214479df98 : FAILURE in 5m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19e0c9f0759043eb9569d6bdd25f632e : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ad8c9f9f8140968124b000ced02c20 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a21a32d7f78740619acc438138051bec : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/003d1070c5d04bf2b82ff6c61c14b47b : FAILURE in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d86267b7ddaf42efaafe581f3dc5d99c : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/206776d03fb34d6b9f3cce50ea789e13 : FAILURE in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06cc0c4c9b714bfb8cc6cbb7b9b614e4 : SUCCESS in 4m 00s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/bacdf65688cc489f831014f8873f004c : FAILURE in 20m 24s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/ec0b314997074ceea4ea89df491d5eac : FAILURE in 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"799a0db85652a877ab7ed95ae74888b68d1ef2fb","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-22 08:37:50.000000000","message":"Patch Set 1: Code-Review-1\n\nWe should hold on in these patches for a while and observe oslo.cache for some time before spreading the change around all oslo.","accounts_in_message":[],"_revision_number":1},{"id":"b59271cf56441718bfab5fb165cbb3d672a8007c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-07-22 08:44:16.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"62adb7585232ed5c94134804b26b1209af3aa8a3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 08:34:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8934c892ab4e00b76895d4b1834c2e1985ff2846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:51:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/619c32975bb04625a73009f1f2e8e6f9 : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7428ea25ab224509bec8afe99304c53c : SUCCESS in 55m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e43b6d61d6b24c4c9c76c2516b746fbf : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0c5874b552b485d85fe2839feac4e9e : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7a8e49df1f74472897ae6dae00b06c2 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c93b8685f640450e872a8e8f807ad451 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bdf9a3db7e74fdaae71eb195843a6fe : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/712cfaa3458b4b6c971f740d55188441 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/713d98b1711e41d497312edc20be1798 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4f4d99cddd4f3a91f5b39f210c56ed : SUCCESS in 4m 44s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/ef291d54373d49bf9f6c74f7ebb77a19 : SUCCESS in 37m 12s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/96b1bff6aa494e79b18d7def468ca5c2 : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":2},{"id":"12a60fcca8b05a650814b4bcc642f8825b6206aa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-15 15:45:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"47e48f81166d847714b34e7f74fbd454c6f3c4c1","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-15 15:46:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a713e5b79b5bd682a7746e546e69cb6eb5423a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 16:27:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/58896507a83f439e884723145b484bb5 : FAILURE in 17m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e3f35b4893e4287a64a60414ffc3225 : FAILURE in 17m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67d9b572c9d74c4586a45772c8828479 : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b383507325d74b6f8edc7fe96c57d94a : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63dfded59d95496c9628c348db8187c7 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3c74badd22b41d9bffd133e3aaf4289 : FAILURE in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59eb293aa4a747309c43e1586b166886 : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8716ad4ef0544679389c425c12f652f : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33079cbeca01447d982779f359616999 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878c02d0e90442e889586b4dcd655e69 : SUCCESS in 5m 34s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/09fc54e0119f4dc585f50eaeb564251b : FAILURE in 27m 17s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/31d9387abebd49bdb48d11cbe101037e : FAILURE in 32m 17s","accounts_in_message":[],"_revision_number":3},{"id":"7d381290ad56b6ba252fa320fd47e65600a1ac75","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-09-16 08:52:55.000000000","message":"Removed Code-Review-2 by Moisés Guimarães \u003cmoguimar@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"e3e4246bd6411b0e142d5c07d61427c8d88794ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-22 15:17:31.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b266709f1c411c860c4a4476f7d2a4d0f215730b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 16:54:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2e9daf65897842b28feb76187ca2d9c9 : SUCCESS in 1h 23m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/178c54655d2d4b6aab0ca183b2358cae : SUCCESS in 1h 15m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cf3c15c5d0044cc9bdd4ff86db118d4 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/664c456f9d2a4369959b5d2e720e591c : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d13d3295340b4d2cbe3db3efba8e73e7 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b10e534c5ff4efe91b9ca2854cb0352 : FAILURE in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60b234caae474324b646828fc96ed1ef : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/017ab837a82d479f9eb1964590f056bd : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/224d7f13fa0f4c4fa2ba8b113c04b5f9 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1860b274311f4a87b2a2e2fea8ae4351 : SUCCESS in 5m 43s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/cd76ee11fe674c088d5fd496ca050bfc : SUCCESS in 40m 00s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/a0260f0a6ca34349b15d68003077df06 : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":3},{"id":"3cf9484b6c9379409d8ff525ae9126d42b45d863","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-23 12:19:07.000000000","message":"Patch Set 3:\n\nThe Python 2.7 seems legitimate, I think it\u0027s not related to these changes but more related to our supported runtimes and our zuul config.","accounts_in_message":[],"_revision_number":3},{"id":"a393b9fa2a9c693d5447a04cc8eedef4389b856f","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-08 14:24:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ba20d27b17073a996d5f0528f831339281eb8274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 15:59:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9a53f54063e54f4fa8d33369d47837df : FAILURE in 30m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad280aac0e154697b12499435446471d : SUCCESS in 1h 23m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa6571a077d148aca07a97e427a4f8c6 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e6cc039980b40fb9d1067497b8b49b3 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ccdeaa887434ba7af44306b8c621665 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c16063213f4e418db445c4be33f2e85d : FAILURE in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35555804b267454b8991d1ed55367f11 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/672b5ed0950a4c6796329be786aeb8d0 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24bc02774f8648afb33bc8b85a54e4a7 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f037e26f6eaa4733942393e72d03f051 : SUCCESS in 3m 29s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/21960bcde71343b1ad427f3c43ea2b19 : SUCCESS in 23m 36s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/87d5fa2fece84e2abb54f1425013b7f4 : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":3},{"id":"efd302a0f9083b00f1249ef346a9f0011cd8cefe","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-09 08:22:46.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d6c43ee28c4b4855075647e1bfde66931ece490c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-09 09:54:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"48deeaa5f79775b060dc5c7eb695eb6f6b36540c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-09 09:56:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4d48b4b3dc7956a952d7e2fc00f707af1a5e8ae4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-09 10:44:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"37ed9026c9e959cd66103463aa2c488c3baf20f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 11:03:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5adee38510094b41a74d8495d2a6425d : FAILURE in 17m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf587134dc3e49ba8952f3dd978ed146 : SUCCESS in 1h 01m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00baee821f5d4cc29b705082b7ff04bd : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38b298022c6b4786881cf9239b92607d : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6c23d6cc0f4ad7b664b733fbbf2bb9 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7808c709860c4715b07f8822f07e7b59 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1e15ff815eb4c838968deab3e7f44e9 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90f06ce151914aaea6c8f60f79a9a2e1 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b96b94dba004be3b686c340a4903ee7 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e6d59986bc4975bc25c62ac90ec977 : SUCCESS in 4m 17s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/6ab790cdc3b74c099d9864c12912d26f : SUCCESS in 24m 40s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/0ea8ce48e657442cbd7d6f753ea5d6e9 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":6},{"id":"b966f8af07147e3473055ed4039163d0acc7c3e1","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-09 15:07:24.000000000","message":"Patch Set 6:\n\nYou\u0027re blocked by : \n\n+ functions-common:apt_get:1111            :   sudo DEBIAN_FRONTEND\u003dnoninteractive http_proxy\u003d https_proxy\u003d no_proxy\u003d apt-get --option Dpkg::Options::\u003d--force-confold --assume-yes install python-guestfs\nReading package lists...\nBuilding dependency tree...\nReading state information...\nPackage python-guestfs is not available, but is referred to by another package.\nThis may mean that the package is missing, has been obsoleted, or\nis only available from another source\nHowever the following packages replace it:\n  python3-guestfs\n\nE: Package \u0027python-guestfs\u0027 has no installation candidate\n\nShould Depends-On : https://review.opendev.org/757127","accounts_in_message":[],"_revision_number":6},{"id":"8e70f68c6b5aec1cf54abc605b8c8ba7f349b369","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-09 15:29:20.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"b6e9ffa6e317d91e148f344acd22dbc3008e1970","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-10-09 15:32:10.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"3d3d9f005c83b4c541e7a254ded3356c493db297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 16:31:35.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2c75bdb5522a4426a94f3b56f3b465aa : FAILURE in 7m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ef0ba2b3da249c788ddd6fd165645ea : FAILURE in 7m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a5baf979d344cd5adc81c65759bbb69 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc4c469b9844458681c314b6030a27a2 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0971feb1a554ecb9ab6df33f939270c : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30fc199ba99a4445950bb8c9d8402de5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac2263bc045a4501b42f3c7b9b2e6684 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee65f03e47e644ab854b433626f1a712 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/086443ce6db7426a9e16443cc2d9c9d8 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545ac155fba54e82baa0d7cf9bf64298 : SUCCESS in 3m 40s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/178c5762c79748feaa1b70da23ab8192 : SUCCESS in 26m 08s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/b3e0ce51692d4cdc84d19476a4d30a3c : SUCCESS in 25m 32s","accounts_in_message":[],"_revision_number":7},{"id":"9c1457e8481267ed3e6647650a6586f1aeaf7cd6","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-10 23:11:03.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"37c6244f23932644628184152f4b13753c26c15f","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-10 23:13:58.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"79587f2f71772b4393abd8013545fd78f0955a26","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-10-10 23:31:36.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f1eaa92d9e78b5dd8ec5673ef435abff78d394d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 00:56:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/379670e5e5424ac882c27274c55a4229 : FAILURE in 1h 38m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/500b369d4dc14ed4bd458b99d5908bfc : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9448a9be8ebb4a2ca8b85f91efe0ddbd : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/917668e30ddd4379aac0e680681b688b : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eaff33d6b78e41a4b822c432c9d3d7cf : FAILURE in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fbee8d8df044f3ba2fc48daf06fa705 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d43bb323a2e4d058a1a14241f068add : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb4ee6d836c0442a87c7e91a0e5211fb : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c51ea8b4c254196a05b6663d5dbe5cf : SUCCESS in 4m 01s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/250c6746284347249445562750048029 : SUCCESS in 25m 43s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/8f91dc8f3980451b88d48e9fbd6d6225 : SUCCESS in 49m 35s","accounts_in_message":[],"_revision_number":9},{"id":"a6994ff653f21debfdaec5998afcb981e9496617","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-12 12:20:20.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"535b3c637aea0e7cc7c27dc354bcf6c6886b590f","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-12 12:23:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b2d6adde83f34af56b8b464a2bd8f8f38e34417e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-12 12:33:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3892465f63de1c6a5a885fdc207ede786b9b66ff","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-12 13:09:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"04b61bf7b76a6eb00c5027247a5b0d03c4ae1c85","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-12 13:28:39.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c0ccfc6da250506deeaee8f7a841409692be94cd","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-12 13:29:17.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"bbb26e93decd6153a069b8310102c9fcee1f1c53","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-12 13:38:37.000000000","message":"Patch Set 9:\n\n(1 comment)\n\n\u003e (1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5d31ac69786ccfd080345bd380fc3da1f7a8f853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 14:50:23.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/593bc024d6684ed3828794cd9aff9d74 : SUCCESS in 1h 15m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc86f21ebdc247e0bb0114e03a4e0de1 : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6659658d189c4ef4bdbc764c42365d06 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2c617b8ffd4943b57f2c53b42c0260 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7af3beeb7e454f3b8dc23b2a9f7f29bd : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4893f0b9386e44969c3f7ab621613c3f : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2aee51789d84aa9a6fe954c39d4acac : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/842fe0781cc24f66872d690b8eb2f7bc : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82fb66f551c4189b93e9b7bb9e537df : SUCCESS in 3m 21s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/baf00aa88e5f420aa310b2f50ffc045f : SUCCESS in 29m 17s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/aa0a7d07c0d2497c8faa9c395e5d3193 : SUCCESS in 27m 13s","accounts_in_message":[],"_revision_number":9},{"id":"6b72620ca37994f170cea8339ebabb27353633bb","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-12 15:21:32.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"0245933891858ef51e375953019a2b6bb4a2515d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 16:59:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/133ab7a80b7c4548ba153596bb4c5052 : FAILURE in 19m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfe37b0b81a144c7b8d2ef0a95941732 : SUCCESS in 1h 19m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8649bc10e4ce49bb864566979096eb4f : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0d0f9dedd824a5e81b1bee7ea894798 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32c9e7ec66b4e3aa9a45687b71b8b2b : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0227fead8a63428b917de5808d95359f : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2e7b14bf44f4e0a9376a779c0d24d66 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f24c9ea67054a1eab271404a6b5ce3a : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2f93a996f654466b9c9f1dc18208001 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dacfdfd2e77f4babb0b124835f7cf35c : SUCCESS in 3m 08s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/b4651f1ac28b489db70ec519cf0e6aa8 : SUCCESS in 23m 01s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/9ba74bc6a9404853b706c001ffba376d : SUCCESS in 26m 27s","accounts_in_message":[],"_revision_number":10},{"id":"c3f7cd71f8a6718c9abb0880c41195ade9ef1919","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-13 09:16:22.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"10f6701611b8c207a8a59c77b423beaebb284d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:48:11.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b49a8c9a3684a3284816a98df820e01 : SUCCESS in 1h 28m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fb755e071d74cf09629f7f05dbdb116 : SUCCESS in 1h 04m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c8c9ab814ab4df6b912b5ad0c7e3780 : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/197755185d3540b091ebb444b70eda7b : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012174e8056a412e9bd4cb60178d1d20 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8d0fc345b1c40d1a6ca12eafa2acfb7 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c621ab3a0327473f8360148de3fbe04f : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca991111c94f4437ad1eb719c91733a7 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3335d5a26fe4d7fba9ce282d855739b : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3a982efb0b40ca98457a88fe8c076c : SUCCESS in 5m 18s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/13b9f98f1b3a48dfbda60b468869fd10 : SUCCESS in 54m 03s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/4fe2b0e0eba94f04a7913f233da27469 : SUCCESS in 48m 39s","accounts_in_message":[],"_revision_number":11},{"id":"3d1e0fa5b8ee0b3f7ed4acbb3206ea8b8af07b31","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-19 15:25:03.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b8e9786cf8d1f76a6edb9b7e986dbde1c1367fb5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-19 16:16:06.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"76e58b4b20e82dea83bf4e4ce6a55da65d04d91c","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-21 16:32:52.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"092b3fb9dc68931ebb28d78e5fce2b7176afb772","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-21 16:34:35.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"df95aceaba0b82cd6a2c80204e223847d52777be","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-21 16:48:08.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"a00a888f095a1e701c82a0008409080c9d42923f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 18:22:48.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9314fe8c0fb84b8d85e8c4bf0f5e03cc : SUCCESS in 1h 27m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e540d9bfb50412d85af6ada7c13d496 : SUCCESS in 1h 16m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c5253515cb9415abe345f19e987eaa2 : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44558fd5a01c47eaaa8fa6dcf4286187 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64528f25161643449a41eca4921a7bfb : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdce4d306a744ba2998611721d3d1b76 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa03e22d0da34b5a8405831b4f198dca : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8cacabe13f304534b8e45b28d5f6fe00 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d99add90733b416f85d3132133e01feb : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20dc3ce9e109402196c18a0f0ef9f0a3 : SUCCESS in 5m 10s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/e5d4fc1d16b349b6b5dbfd276467e509 : SUCCESS in 30m 04s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/5d4e2c19a3f24ca3b754e36054de1be8 : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":13},{"id":"1b9b16e189a1f0ba4b8389f57c586a661d0878e5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-10-21 21:32:54.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"d7a3336ce0d4756d84d99cf74625e5b121576e64","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-21 22:33:58.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"cb170f3addf7d5192e407b548e83a26096bc0024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 22:34:09.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"4be289aac385a2812899c8f5fce53e1f7ff704dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 00:24:14.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a36a3f23c5fc4f0daf24da2b65436eb4 : SUCCESS in 1h 26m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f158390f0ccc4726aee93b83ad6388f7 : SUCCESS in 1h 19m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ea2054a3a7c4128ab7ffc2ae3a7232a : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e0efc88b384830b56597d6a6c8c1a2 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3f3741cde8942d88a0c5bb35aa22d39 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cbdf92d32e64c51be1957afa4743196 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c4e4fe7eb6d4bf7ac8de3c19a5d2fa4 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06eb0ab3487462caec6d71d37bbbc93 : SUCCESS in 6m 18s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/98a49c8119734dcba059624f6c199840 : SUCCESS in 30m 10s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/4ad54dd601794525b3ced94431a39202 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":13},{"id":"619f91309ed8606924d6c77f723e425c659e574f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 00:24:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"d28dfa4d7c5a793094a75149d26faa41e665704e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 00:26:37.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba0ea2738cc242618bddb8783332d358 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"f5e4c6577a92eb97056b89820e6ad19eb9527d8a","revisions":{"a3bb2b70bd176d04acc4a7f7d2216a534717d110":{"kind":"REWORK","_number":1,"created":"2020-07-21 11:44:33.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/60/742160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/1"}}},"commit":{"parents":[{"commit":"9257b6843568052be1dc1a5f3df6f629795e08b2","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9257b6843568052be1dc1a5f3df6f629795e08b2"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"subject":"Add pre-commit","message":"Add pre-commit\n\npre-commit will be used to:\n- trailing whitespace [1].\n- Replaces or checks mixed line ending (mixed-line-ending) [1].\n- Forbid files which have a UTF-8 byte-order\n  marker (check-byte-order-marker) [1].\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs) [1].\n- Check for files that contain merge conflict\n  strings (check-merge-conflict) [1].\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements) [1].\n- Attempts to load all yaml files to verify syntax (check-yaml) [1].\n- replace tabs by space (4 by default)(remove-tabs) [2].\n- Run flake8 checks (flake8) (local)\n\n[2] https://github.com/pre-commit/pre-commit-hooks\n[3] https://github.com/Lucas-C/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a3bb2b70bd176d04acc4a7f7d2216a534717d110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a3bb2b70bd176d04acc4a7f7d2216a534717d110"}]},"branch":"refs/heads/master"},"d429eacd8d2c6132cfaaa2614ebb676e74127831":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-06 08:34:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/60/742160/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/2"}}},"commit":{"parents":[{"commit":"206b795517a13d759f60ed9605eb03ca785ce10a","subject":"Update hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/206b795517a13d759f60ed9605eb03ca785ce10a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-06 08:34:37.000000000","tz":0},"subject":"Add pre-commit","message":"Add pre-commit\n\npre-commit will be used to:\n- trailing whitespace [1].\n- Replaces or checks mixed line ending (mixed-line-ending) [1].\n- Forbid files which have a UTF-8 byte-order\n  marker (check-byte-order-marker) [1].\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs) [1].\n- Check for files that contain merge conflict\n  strings (check-merge-conflict) [1].\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements) [1].\n- Attempts to load all yaml files to verify syntax (check-yaml) [1].\n- replace tabs by space (4 by default)(remove-tabs) [2].\n- Run flake8 checks (flake8) (local)\n\n[2] https://github.com/pre-commit/pre-commit-hooks\n[3] https://github.com/Lucas-C/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d429eacd8d2c6132cfaaa2614ebb676e74127831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d429eacd8d2c6132cfaaa2614ebb676e74127831"}]},"branch":"refs/heads/master"},"f2df09a06bf5cefbf216d2a94bbd76509e605397":{"kind":"REWORK","_number":3,"created":"2020-09-15 15:45:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/60/742160/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/3"}}},"commit":{"parents":[{"commit":"206b795517a13d759f60ed9605eb03ca785ce10a","subject":"Update hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/206b795517a13d759f60ed9605eb03ca785ce10a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-09-15 15:45:15.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f2df09a06bf5cefbf216d2a94bbd76509e605397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f2df09a06bf5cefbf216d2a94bbd76509e605397"}]},"branch":"refs/heads/master"},"ac88659f0604c1b7fe51a367e83a0184d13b97b2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-10-09 08:22:46.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/4"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-10-09 08:22:46.000000000","tz":0},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ac88659f0604c1b7fe51a367e83a0184d13b97b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ac88659f0604c1b7fe51a367e83a0184d13b97b2"}]},"branch":"refs/heads/master"},"2b0bf27e1aaa7ebb45bb419a7705219bc232af8e":{"kind":"REWORK","_number":5,"created":"2020-10-09 09:54:55.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/5"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-10-09 09:54:51.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2b0bf27e1aaa7ebb45bb419a7705219bc232af8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2b0bf27e1aaa7ebb45bb419a7705219bc232af8e"}]},"branch":"refs/heads/master"},"0b11a29df0ff1dbd435e4ee90f31591f39a0bdb3":{"kind":"REWORK","_number":6,"created":"2020-10-09 09:56:46.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/6"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-10-09 09:56:38.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0b11a29df0ff1dbd435e4ee90f31591f39a0bdb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0b11a29df0ff1dbd435e4ee90f31591f39a0bdb3"}]},"branch":"refs/heads/master"},"b64d6dacf4a70f40473f88c258c4070364709f9c":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-10-09 15:29:20.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/7"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-10-09 15:29:11.000000000","tz":0},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nDepends-On: https://review.opendev.org/757127\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b64d6dacf4a70f40473f88c258c4070364709f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b64d6dacf4a70f40473f88c258c4070364709f9c"}]},"branch":"refs/heads/master"},"8bc6739bf22a9122149c0fe354fb398d94c8d2c2":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-10-10 23:11:03.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/8"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-10-10 23:10:57.000000000","tz":0},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8bc6739bf22a9122149c0fe354fb398d94c8d2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8bc6739bf22a9122149c0fe354fb398d94c8d2c2"}]},"branch":"refs/heads/master"},"1bbb5b10ecb099e28941687956384855cba37cd5":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-10-10 23:13:58.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/9"}}},"commit":{"parents":[{"commit":"6b2bbbd8b3072bd5d14c3d62a151c0a986142067","subject":"drop tempest-full","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6b2bbbd8b3072bd5d14c3d62a151c0a986142067"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-10-10 23:13:31.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1bbb5b10ecb099e28941687956384855cba37cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1bbb5b10ecb099e28941687956384855cba37cd5"}]},"branch":"refs/heads/master"},"e86f7eec5f9bff21cfe1383ba4c61aff3cecbe4a":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-10-12 15:21:32.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/10"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-10-12 15:21:15.000000000","tz":120},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e86f7eec5f9bff21cfe1383ba4c61aff3cecbe4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e86f7eec5f9bff21cfe1383ba4c61aff3cecbe4a"}]},"branch":"refs/heads/master"},"f1389af85ba16aef08d82a1ff19b5f8865673d6f":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2020-10-13 09:16:22.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/11"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-10-13 09:16:18.000000000","tz":0},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nDepends-On: https://review.opendev.org/757127\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f1389af85ba16aef08d82a1ff19b5f8865673d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f1389af85ba16aef08d82a1ff19b5f8865673d6f"}]},"branch":"refs/heads/master"},"34f509c593f35f3a61dd6f5620e418c65178fe24":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2020-10-21 16:34:35.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/12"}}},"commit":{"parents":[{"commit":"16f19608c446db67e2ad0950a3209078ec6be18b","subject":"Merge \"trivial: Improve logging of run commands\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/16f19608c446db67e2ad0950a3209078ec6be18b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-10-21 16:34:24.000000000","tz":0},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/34f509c593f35f3a61dd6f5620e418c65178fe24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/34f509c593f35f3a61dd6f5620e418c65178fe24"}]},"branch":"refs/heads/master"},"f5e4c6577a92eb97056b89820e6ad19eb9527d8a":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-10-21 16:48:08.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/60/742160/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/60/742160/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/60/742160/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/60/742160/13"}}},"commit":{"parents":[{"commit":"5150198c294802f82ddc7ed5765546653bc19adf","subject":"Merge \"Remove use_2to3 backward compat for Setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5150198c294802f82ddc7ed5765546653bc19adf"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-21 11:44:18.000000000","tz":120},"committer":{"name":"Moisés Guimarães","email":"moguimar@redhat.com","date":"2020-10-21 16:48:08.000000000","tz":0},"subject":"Adding pre-commit","message":"Adding pre-commit\n\nIntroduced changes:\n- pre-commit config and rules\n- Add pre-commit to pep8 gate, Flake8 is covered in the pre-commit hooks.\n- Applying fixes for pre-commit compliance in all code.\n\nAlso commit hash will be used instead of version tags in pre-commit to\nprevend arbitrary code from running in developer\u0027s machines.\n\npre-commit will be used to:\n- trailing whitespace;\n- Replaces or checks mixed line ending (mixed-line-ending);\n- Forbid files which have a UTF-8 byte-order marker (check-byte-order-marker);\n- Checks that non-binary executables have a proper\n  shebang (check-executables-have-shebangs);\n- Check for files that contain merge conflict strings (check-merge-conflict);\n- Check for debugger imports and py37+ breakpoint()\n  calls in python source (debug-statements);\n- Attempts to load all yaml files to verify syntax (check-yaml);\n- Run flake8 checks (flake8) (local)\n\nFor further details about tests please refer to:\nhttps://github.com/pre-commit/pre-commit-hooks\n\nChange-Id: I730018e2bf788be5ac4e993af1746296206edfc1\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f5e4c6577a92eb97056b89820e6ad19eb9527d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f5e4c6577a92eb97056b89820e6ad19eb9527d8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
