)]}'
{"id":"openstack%2Fpbr~744335","triplet_id":"openstack%2Fpbr~master~I936dcaaf432377d56559bfb627a03772d26d26d5","project":"openstack/pbr","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I936dcaaf432377d56559bfb627a03772d26d26d5","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-01 23:07:37.000000000","updated":"2020-08-17 00:04:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2134476e79a5fd9f1de28f1acc8efe6a1d5e035c","_number":744335,"virtual_id_number":744335,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-16 22:08:13.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-06 10:21:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 10:21:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51790e8d1f55c5e02106f59c24b97aa28987cb7a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 23:07:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b9f68203bee7fc35f2a6af18e7d4b8b25b0e8c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 00:17:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/70dbfc408dbc407fbc825a1a83302c76 : SUCCESS in 54m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef51c55a8b584ec182a0d48f62ffa27c : SUCCESS in 1h 04m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea2f412b5eb14d3cab46a888613d274b : FAILURE in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98f1ebcb1e8a4e209bfef4c86b14c21e : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68eb07d3bf334808a234e7832e756134 : SUCCESS in 2m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eba0a2192af641b7a2713c65760cb0b8 : FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19309ddc07ac4ce79991812f13ecb91b : FAILURE in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b1799baa066456680538c13c4ab1b5a : FAILURE in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d4e8049043449d92c9018db9b323f0 : FAILURE in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7f8ef4d07642498c15e565014e17de : SUCCESS in 3m 52s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/ec05727dfedb49e59a9c333bbbf2428d : FAILURE in 12m 30s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/fe2cd1f4acee44d5a62ffa254feb8ed5 : FAILURE in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bee1bf8df2caf9b27363794e578b9ca5ec124179","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:55:56.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f5d21c844cb8180f91a8cab478862d479c18460a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 21:46:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d8dbdccb0184ab499c44c5fca95930a : FAILURE in 12m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59cad20cfbb14264a81e4b9831a1f903 : SUCCESS in 1h 15m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/902724a963814c6283c18132172cf365 : FAILURE in 4m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eaf341a140344fdbaeb5f4084a7d273c : FAILURE in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0302a8bd16453ea988f000aa041fec : FAILURE in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f4698394cad403b8a25cbb9215b74a4 : FAILURE in 2m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f3a3074d785475f821ae0b55eadb468 : FAILURE in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/392dcfac30c042fab63ddac0ac413531 : FAILURE in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/994440acd1114b04b245c8f0675ab197 : FAILURE in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1970464a054b958c2fe16c786aa5fc : SUCCESS in 3m 55s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/c7708693cc8044bcb3aeb3d3de89083e : FAILURE in 15m 03s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/ad4f557903fc49afa829c395b57d1df6 : FAILURE in 11m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8188e5a5532b609d1a99a040abff8242d471047b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 08:32:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1c6657aa5fe34fcfe8c9dead367b66cdaa74cfff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 10:21:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f483634bf1f54779b1d982cb9885d30f : FAILURE in 27m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5ecb34032e7433b8a88c239900a3077 : SUCCESS in 1h 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35205c9a55d44c83b01613c181b83259 : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2183a49679ca4a4084b4d683bacc136e : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061fb332698d47bca7f3e33a4d8c4895 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5f5f5f1d00d438d96ade0462a868a5d : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/257d276118c041beb2e40498f3d5ce5f : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afdffa04a1e046b78d2149fab10fbe60 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff390de9ed38488a83d9070dfc267038 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17234efd5ea34b90b69cb6b37c81dce0 : SUCCESS in 5m 21s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/b3fa0a022b0945848fe5b36f21560769 : SUCCESS in 34m 05s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/3879d6316dd744c098be5677b7837273 : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":3},{"id":"8acc42e393b949d41f711199328e2290156f0ae9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-16 22:08:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"23f7985979ea871622bc1a129ec53b809bf72fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 23:16:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/36dc4279848944e6acece1da35acc6ce : FAILURE in 15m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8a5fb23247644608d937d43144373b6 : FAILURE in 1h 04m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f18d3a6c1af411fa4b16792efd83d79 : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4501c679476b41f39ecf1d965f0143e6 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1219270c2c814234a1b75a819bcf49c7 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e640aeb76f914b7fbc63acf754692c58 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebef3c4a8ab543589e1974e81cf1b132 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41f0684277b84b229ad2f40860f8a550 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23a9e136b5ae45e1a560a709bd65e196 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afdba92c7b4441f3908e0dc9ba828252 : SUCCESS in 3m 26s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/24bff5dd80ec4b868939e319f7ca4a49 : SUCCESS in 27m 32s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/37647b3c1959495eb01c0a092bbe2e80 : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ff21e8896629a5b905ff8cdb18d45509d9845cdd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-17 00:04:28.000000000","message":"Abandoned\n\nno related failure so no work needed here for Focal migration","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"68caa42ed3d946cb6ae72aab6a6cc8b1ba67328d","revisions":{"af3010bbc4bc4dce1c6534d7b65a50394d8abaa3":{"kind":"REWORK","_number":1,"created":"2020-08-01 23:07:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/744335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/35/744335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/35/744335/1"}}},"commit":{"parents":[{"commit":"9257b6843568052be1dc1a5f3df6f629795e08b2","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9257b6843568052be1dc1a5f3df6f629795e08b2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:06:19.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:07:36.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nChange-Id: I936dcaaf432377d56559bfb627a03772d26d26d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/af3010bbc4bc4dce1c6534d7b65a50394d8abaa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/af3010bbc4bc4dce1c6534d7b65a50394d8abaa3"}]},"branch":"refs/heads/master"},"be6c6f038ea9f2527f687464915713b3406935a6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-02 19:55:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/744335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/35/744335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/35/744335/2"}}},"commit":{"parents":[{"commit":"9257b6843568052be1dc1a5f3df6f629795e08b2","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9257b6843568052be1dc1a5f3df6f629795e08b2"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:06:19.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 19:55:53.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I936dcaaf432377d56559bfb627a03772d26d26d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/be6c6f038ea9f2527f687464915713b3406935a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/be6c6f038ea9f2527f687464915713b3406935a6"}]},"branch":"refs/heads/master"},"68caa42ed3d946cb6ae72aab6a6cc8b1ba67328d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-06 08:32:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/744335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/35/744335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/35/744335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/35/744335/3"}}},"commit":{"parents":[{"commit":"206b795517a13d759f60ed9605eb03ca785ce10a","subject":"Update hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/206b795517a13d759f60ed9605eb03ca785ce10a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:06:19.000000000","tz":-300},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-06 08:32:36.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I936dcaaf432377d56559bfb627a03772d26d26d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/68caa42ed3d946cb6ae72aab6a6cc8b1ba67328d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/68caa42ed3d946cb6ae72aab6a6cc8b1ba67328d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
