)]}'
{"id":"openstack%2Fpbr~744704","triplet_id":"openstack%2Fpbr~master~I56b0e2b86a7fc66c5ae25e06175033a885f500f6","project":"openstack/pbr","branch":"master","hashtags":[],"change_id":"I56b0e2b86a7fc66c5ae25e06175033a885f500f6","subject":"Update some url to use opendev.org","status":"MERGED","created":"2020-08-04 13:17:05.000000000","updated":"2020-08-06 16:26:19.000000000","submitted":"2020-08-06 16:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744704-1596731033481-88ec3396","meta_rev_id":"fd0551ca658d55b39e2147727dede199643bb2a1","_number":744704,"virtual_id_number":744704,"owner":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 16:23:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-06 14:18:12.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-08-06 13:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-06 14:18:12.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-08-06 13:10:19.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 13:21:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-08-06 13:10:19.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-06 13:14:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-08-06 14:18:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-06 16:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9ab44f967b6ea01742c7af674594889c07c6e72","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2020-08-04 13:17:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c00251042194cb2e692735ccc3890119da5f7639","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-08-04 13:21:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9ec85711092733235ccdfc28b8a712352742e1e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-08-04 14:02:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06207b9fae7194b3d9483bac157f2dcc36526b8c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-04 14:06:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c28e9b6a368a60ec72e7481f598eec38448d2ec2","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2020-08-04 14:18:43.000000000","message":"Patch Set 1:\n\nThank you Jeremy for the follow up patch :)","accounts_in_message":[],"_revision_number":1},{"id":"951177273fcfbc254cbbc761227bbf36048854d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:11:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32783836d6654d4193248d5d258d80ee : SUCCESS in 4m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b03684dff244acea90d4f10985ea1e2 : SUCCESS in 1h 43m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21fa6cd719364e4a9fa2112837450dbb : SUCCESS in 1h 32m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edf2c6fc10f248aba899a0e630ec46c9 : FAILURE in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/877ae52d013949a3bb330a948a4e63e5 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bcd56ddd0054bf490ca2f395ffbe6d6 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/63f6b8f3bf51492881e591d6d16320f4 : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8deed9e1af0640f1adae79900fbdc86a : FAILURE in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00ac82ba637843e1bb4be30bf5c013ee : FAILURE in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebdf91562c824099b42262c95b8531e6 : FAILURE in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710326f4f36f4183b9d8f60795338caf : SUCCESS in 5m 23s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/a810697083764c1b8dff7ec70394b3a3 : FAILURE in 19m 32s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/b17193de1ca64787874fd906bd4f0f50 : FAILURE in 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"103a7b70e759ea5ea600181f0a4063c990b04c21","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 08:30:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cc0f4ebdc0becdeeade96b6d7e179d34daa888bd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 08:30:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8966005c004b379cb7bb9f4a7cefd7e93ce03443","author":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"date":"2020-08-06 08:48:46.000000000","message":"Patch Set 2:\n\npbr.tests.test_packaging.TestRequirementParsing.test_requirement_parsingtesttools.testresult.real._String fails with:\n\n\n  File \"/home/zuul/src/opendev.org/openstack/pbr/.tox/py37/lib/python3.7/site-packages/fixtures/fixture.py\", line 197, in setUp\n    self._setUp()\n  File \"/home/zuul/src/opendev.org/openstack/pbr/pbr/tests/test_packaging.py\", line 186, in _setUp\n    virtualenv.create_environment(path, clear\u003dTrue)\n AttributeError: module \u0027virtualenv\u0027 has no attribute \u0027create_environment\u0027\n\nThe test creates a virtualenv inside the test virtualenv:\n\n    def _setUp(self):\n        path \u003d self.useFixture(fixtures.TempDir()).path\n        virtualenv.create_environment(path, clear\u003dTrue)\n\nAnd abandoned change https://review.opendev.org/#/c/744793/ states:\n\n\u003e Drop py27 support and fix compatiblity with virtualenv 20.x+\n\u003e \n\u003e This major release of virtualenv has some breaking changes. The\n\u003e create_environment method has been removed and the recommended\n\u003e replacement is to call\n\u003e run_cli.\n\n\nSo looks like an issue with virtualenv / the test environment rather than with this patch :\\","accounts_in_message":[],"_revision_number":2},{"id":"58c6043e5cbda62c37b4731ce8e8f5c5e59db4c9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 08:58:35.000000000","message":"Patch Set 2:\n\n\u003e pbr.tests.test_packaging.TestRequirementParsing.test_requirement_parsingtesttools.testresult.real._String\n \u003e fails with:\n \u003e \n \u003e \n \u003e File \"/home/zuul/src/opendev.org/openstack/pbr/.tox/py37/lib/python3.7/site-packages/fixtures/fixture.py\",\n \u003e line 197, in setUp\n \u003e self._setUp()\n \u003e File \"/home/zuul/src/opendev.org/openstack/pbr/pbr/tests/test_packaging.py\",\n \u003e line 186, in _setUp\n \u003e virtualenv.create_environment(path, clear\u003dTrue)\n \u003e AttributeError: module \u0027virtualenv\u0027 has no attribute\n \u003e \u0027create_environment\u0027\n \u003e \n \u003e The test creates a virtualenv inside the test virtualenv:\n \u003e \n \u003e def _setUp(self):\n \u003e path \u003d self.useFixture(fixtures.TempDir()).path\n \u003e virtualenv.create_environment(path, clear\u003dTrue)\n \u003e \n \u003e And abandoned change https://review.opendev.org/#/c/744793/ states:\n \u003e \n \u003e \u003e Drop py27 support and fix compatiblity with virtualenv 20.x+\n \u003e \u003e\n \u003e \u003e This major release of virtualenv has some breaking changes. The\n \u003e \u003e create_environment method has been removed and the recommended\n \u003e \u003e replacement is to call\n \u003e \u003e run_cli.\n \u003e \n \u003e \n \u003e So looks like an issue with virtualenv / the test environment\n \u003e rather than with this patch :\\\n\nThat\u0027s resolved with https://review.opendev.org/#/c/719964/","accounts_in_message":[],"_revision_number":2},{"id":"576893bb51e7f69df72ae78e66cc5fc19bfcfbcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 10:02:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d1dd539cf8154347bd48ff1dd428d07b : SUCCESS in 3m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fdca71d1525045759f6f8bd51531d3b9 : SUCCESS in 1h 22m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4b9afbee2db490eaa68e3de0186b5cf : SUCCESS in 1h 23m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0545f06075fd42f3bd17822100769c4b : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86d46c8122a04f18bea7f52bd433ae1d : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71881b38a60c41a6a02b8c1ded1215cd : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/228053d7fe514f35af6a223d17bf3a84 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eecaf49322f24d089de056d874e8bd76 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fdbb6b973304849ad5b1b5af411d4fc : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d91bc50c1f1c4757a3b13e6673e12cad : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbac83eefd44b30adfb0b0629504c2a : SUCCESS in 4m 07s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/f9fd11c8d55640ebb3586258094dbf7b : SUCCESS in 43m 55s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/2f7d95a785f0455195b5ce96d259557e : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":2},{"id":"14cbc1bfa2ba8a8cd3c700a0e2973da62ed5c4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 10:02:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6db73940ae16926a0cdb22d9be2b397b87e468c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 12:09:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/69f816eb116f49f381fe75a3125d358e : SUCCESS in 4m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c415a28741a048738839758e7349725e : SUCCESS in 1h 35m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/503e0afe39774172bce337956bfffb08 : SUCCESS in 1h 31m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/672f41a171fa49d5994c4fd0b7e7902b : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5703b2096adb4d22b7a99cdc793802ac : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b3213f86d9b4cdab8461a77c5e4fc25 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d0a46e654b1492b87ee029348da29f0 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ffaf8c8ff544b098e9c8a4d05be0aea : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64bb3177dab48c0b25d832b59d737aa : SUCCESS in 4m 28s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/8d7738aad8c54dcd9d5dd0f3601dab00 : FAILURE in 43m 21s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/294502dde3004f6e95ba589a34a28d03 : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":2},{"id":"46287ad58e432101b437df2c93d92c9330b38d48","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-08-06 12:51:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"10a1ffcaa9abf933fb499015204bdbdb6c8d9956","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-06 13:10:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1767d2658a6cae92096e949f21d64cd1a38c3a7f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-06 13:10:25.000000000","message":"Patch Set 2:\n\nI\u0027ve rebased this.","accounts_in_message":[],"_revision_number":2},{"id":"7da43047ca5645c3fd4df925a279dcc9bbecd04f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-06 13:14:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"991e8c16b36b2e66db4b6d5bac6a854c4ee142f9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-06 14:18:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0ccbe88f8c182da69f334b7c505bcedeb088452c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 14:57:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c044f90a5d64a0b948d970e9f75f40b : SUCCESS in 2m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4de18b9a94e94d999291ffb843ec04f6 : SUCCESS in 1h 44m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f049850e67ec4416906b63ea2bf230b3 : SUCCESS in 1h 16m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7370225ecd7947e1bdcf99eb21857348 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bad87ef85dc4a86853c599cc58b388d : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3d8e7379f14946b4dd6ef336028043 : SUCCESS in 2m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75f23557217f4459847a811d600d7685 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab68820667e64519abd32583b1070489 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ddd8d3e6cfc41948edb24150bb50220 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97dfa5c112694b819158d01636a9fcc4 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bcaaa1d4de243a7addf97ce643d395d : SUCCESS in 3m 30s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/43359d41b9d04957a99ddc127dd1d1ca : SUCCESS in 30m 08s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/caeda2bcff624feb80dd851badedea92 : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":3},{"id":"d3aad3a25207154de9fbe756e46e6dddf047de5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 14:57:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3d5dabd21c294547be81d396b8f477df893e7e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:23:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f4e0c144b67e45f79674b844afecbb29 : SUCCESS in 2m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3eb3964f6db54dea9402f150dad34a9a : SUCCESS in 1h 22m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20a8f566a5f244f791afcc6cbaee0827 : SUCCESS in 1h 16m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc321af6caca4b96bff62cc6e0fcb269 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a503b0e94246439ca9073b9b537d02d7 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f6e7db58cda4352982a710eaf7c0914 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c2b4ea61f1445aba5b792e750dccea2 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/845c3ab5b6ec4b8d83a27789598d7b83 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe2d313d39b4b24941d8771da1254c7 : SUCCESS in 4m 28s\n- pbr-installation-devstack https://zuul.opendev.org/t/openstack/build/949e31fde6934dc7adeb3b038046373e : SUCCESS in 32m 28s\n- pbr-installation-upstream-devstack https://zuul.opendev.org/t/openstack/build/842de6a465f542dabfffcf7e89d90846 : SUCCESS in 25m 07s","accounts_in_message":[],"_revision_number":3},{"id":"a43ec01404fb40b40c45630df038a8c74f9e246b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:23:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fd0551ca658d55b39e2147727dede199643bb2a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:26:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4606b3704bad489ba35aef62831cfbb4 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"07b6ac85e65c6ea104945a01dbb0622312a2ed56","revisions":{"ff2249d8d02055fb577cdb5e3a5752be1d284e35":{"kind":"REWORK","_number":1,"created":"2020-08-04 13:17:05.000000000","uploader":{"_account_id":2475,"name":"Antoine Musso","email":"hashar@free.fr","username":"hashar"},"ref":"refs/changes/04/744704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/04/744704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/04/744704/1"}}},"commit":{"parents":[{"commit":"9257b6843568052be1dc1a5f3df6f629795e08b2","subject":"Merge \"Switch to Ussuri jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9257b6843568052be1dc1a5f3df6f629795e08b2"}]}],"author":{"name":"Antoine Musso","email":"hashar@free.fr","date":"2020-08-04 13:15:00.000000000","tz":120},"committer":{"name":"Antoine Musso","email":"hashar@free.fr","date":"2020-08-04 13:16:58.000000000","tz":120},"subject":"Update some url to use opendev.org","message":"Update some url to use opendev.org\n\nThe git web URL is now: https://opendev.org/openstack/pbr\n\nAlso stop linking README.rst to no more existing history/changelog\ndocumentation page (removed by 96b84a90141).\n\nChange-Id: I56b0e2b86a7fc66c5ae25e06175033a885f500f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ff2249d8d02055fb577cdb5e3a5752be1d284e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ff2249d8d02055fb577cdb5e3a5752be1d284e35"}]},"branch":"refs/heads/master"},"395196ae4df44fe235e150654fb84dff31ca1bde":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-06 08:30:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/744704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/04/744704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/04/744704/2"}}},"commit":{"parents":[{"commit":"206b795517a13d759f60ed9605eb03ca785ce10a","subject":"Update hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/206b795517a13d759f60ed9605eb03ca785ce10a"}]}],"author":{"name":"Antoine Musso","email":"hashar@free.fr","date":"2020-08-04 13:15:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-08-06 08:30:17.000000000","tz":0},"subject":"Update some url to use opendev.org","message":"Update some url to use opendev.org\n\nThe git web URL is now: https://opendev.org/openstack/pbr\n\nAlso stop linking README.rst to no more existing history/changelog\ndocumentation page (removed by 96b84a90141).\n\nChange-Id: I56b0e2b86a7fc66c5ae25e06175033a885f500f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/395196ae4df44fe235e150654fb84dff31ca1bde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/395196ae4df44fe235e150654fb84dff31ca1bde"}]},"branch":"refs/heads/master"},"07b6ac85e65c6ea104945a01dbb0622312a2ed56":{"kind":"REWORK","_number":3,"created":"2020-08-06 13:10:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/04/744704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/04/744704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/04/744704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/04/744704/3"}}},"commit":{"parents":[{"commit":"3f896a8d846de0682e2bfa5a8b45fc235620c7d5","subject":"Merge \"Remove neutron-fwaas from the jobs\u0027 required project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f896a8d846de0682e2bfa5a8b45fc235620c7d5"}]}],"author":{"name":"Antoine Musso","email":"hashar@free.fr","date":"2020-08-04 13:15:00.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-08-06 13:09:50.000000000","tz":120},"subject":"Update some url to use opendev.org","message":"Update some url to use opendev.org\n\nThe git web URL is now: https://opendev.org/openstack/pbr\n\nAlso stop linking README.rst to no more existing history/changelog\ndocumentation page (removed by 96b84a90141).\n\nChange-Id: I56b0e2b86a7fc66c5ae25e06175033a885f500f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/07b6ac85e65c6ea104945a01dbb0622312a2ed56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/07b6ac85e65c6ea104945a01dbb0622312a2ed56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
