)]}'
{"id":"openstack%2Fpbr~745171","triplet_id":"openstack%2Fpbr~master~I77a38c4611dc28db79d6f25d96ad0de36b224c98","project":"openstack/pbr","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I77a38c4611dc28db79d6f25d96ad0de36b224c98","subject":"Begin work to modernize pbr\u0027s integration testing","status":"MERGED","created":"2020-08-06 15:35:18.000000000","updated":"2020-08-10 19:40:31.000000000","submitted":"2020-08-10 19:37:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745171-1597088247306-ea02c699","meta_rev_id":"2ee194e3df82cb154b4e8d7f095ec66288ca7184","_number":745171,"virtual_id_number":745171,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 19:37:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-10 15:19:41.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-10 17:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 17:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 15:19:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-10 17:48:03.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-08-10 19:37:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a8a28d68cd5e8f7949ede300b3e9cc22587aa30","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-06 15:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87efde009a4332e4057dbe47e0b8987bcc98fbaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 15:35:36.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job pbr-installation-devstack not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/pbr\n      templates:\n        - official-openstack-repo-jobs\n        - publish-to-pypi\n      periodic:\n        jobs:\n          - pbr-installation-devstack:\n              branches: master\n          - pbr-installation-upstream-devstack:\n              branches: master\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4504, column 3","accounts_in_message":[],"_revision_number":1},{"id":"cb13a46cf3b109a16b7c2aaebedc166b0e46b9a4","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-06 15:43:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bc7c9f02a19ab7fdc93f8156dcf28183097cfbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 15:59:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/2e32f5da68284ff7826e60c88329c5d5 : FAILURE in 10m 43s","accounts_in_message":[],"_revision_number":2},{"id":"94c31e01d67b9124674e9c9ee15a2618e8a4761a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-06 16:03:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2f83515087f2c71716feaa4006a5b115452e445a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:17:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/1bf058ee421e47bb963eb0dd3fe09703 : FAILURE in 10m 00s","accounts_in_message":[],"_revision_number":3},{"id":"d2df7697e4451e13c0a80405b341abf55a83d0ec","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-06 16:18:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e4cb5f65342a661bc8b420478a7ecaf73e596b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:51:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/3073ce080598407fa5978a7136e6040d : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":4},{"id":"d250c96ed39c9d22ee7c8d35971733df3cb222cc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-06 16:58:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"291ae664203b202034c0c2c2dfb8cbfe359245b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 17:45:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/23ede5b01d9a4c57a7776091be7d8de7 : SUCCESS in 42m 12s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/1e9c88588dd941a28bdc2e905ae45b89 : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":5},{"id":"0c05c282af43e4966f943fa952758a9cfa5b9640","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-06 17:46:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a8d167cace7433e5f9892c7068d4c6132e15749a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:14:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c3b94d7df4244b6b26ce7da7e234491 : SUCCESS in 1h 11m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc66103e6383474494db6e47d1242d5d : SUCCESS in 1h 22m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96364ed7f2d84583b01f61efe0c756aa : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb40c538f1454ed3ab8d994180841141 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d49380833734421ea43554006204a74f : SUCCESS in 2m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ac08217a8fa4418a54c01249d4343de : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bf51ebeb5334f3ab8f7728acb0f5aec : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb43bcdc7c2a47e3abef2d43e290e0ab : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b129ee4f58c345218b4c76d08e19a55d : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399206bba759405ea1826d0f52acf95b : SUCCESS in 4m 15s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a5d42cc9125d4715b2137ec9f4913ead : SUCCESS in 23m 46s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/52ba537c22e14dd5873ed7388c3b7ff8 : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":6},{"id":"26962803bcba5c7ff6e41a41b05be39b998e7d59","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-10 14:17:18.000000000","message":"Topic changed from modernize-integration-testing to native-zuulv3-migration","accounts_in_message":[],"_revision_number":6},{"id":"8f623454829659abba0daf84de52f602b4da1c3a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-10 15:19:41.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"da57193d197abce423249b42622cc4889a505a06","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-08-10 17:48:03.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d1f8b31376144eaaa563ea7b23c2183e7dee419d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:48:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d0fa39b59b5aae377a8f335f98068f9f5dd025a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:37:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/59f79a3b1cef42318aa03dc6bc15f649 : SUCCESS in 1h 28m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35c4d03592ed49e7aaeef91a04784af9 : SUCCESS in 1h 42m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc1f298cabc54ac8a2eecb3e4b8fb920 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fe166f9607a458686a3763f39c0eb4e : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f7720c804f4426aab403599dcfc3dbf : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e3ea5e0f87f4bc89aa8e045265e2ac8 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73673f896e8a471a914b20b1181913ba : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916a1758986743f28a9db583892281bd : SUCCESS in 5m 11s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/ee6920193f4f49ffb9f02176dff88216 : SUCCESS in 30m 14s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/386f578e39ac4ee48972e9d24ba5a662 : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":6},{"id":"2ccd86818b8cd7c3749b03507c05ee17155f4f0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:37:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"2ee194e3df82cb154b4e8d7f095ec66288ca7184","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:40:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e76a57e94d14a0b9275c2e812c08bfd : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9e229269b208671af5f4f3431b3f077dda5453c5","revisions":{"d4a62f3348b36f4315d9e959910c4e4a4592fbfd":{"kind":"REWORK","_number":1,"created":"2020-08-06 15:35:18.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/71/745171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/71/745171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/71/745171/1"}}},"commit":{"parents":[{"commit":"3f896a8d846de0682e2bfa5a8b45fc235620c7d5","subject":"Merge \"Remove neutron-fwaas from the jobs\u0027 required project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f896a8d846de0682e2bfa5a8b45fc235620c7d5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:33:40.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:33:40.000000000","tz":-420},"subject":"WIP: Begin work to modernize pbr\u0027s integration testing","message":"WIP: Begin work to modernize pbr\u0027s integration testing\n\nPBR does integration testing by installing all of openstack\u0027s python\nprojects to ensure we can install all of them in a variety of different\nways with different pip versions. Unfortunately, these jobs are still\nusing devstack-gate. We can simplify them significantly because zuul is\nnow doing what devstack-gate does and provides repos configured the way\nwe want them to be.\n\nThis change update the jobs to drop devstack-gate and simply consume\nzuul\u0027s repo prep.\n\nChange-Id: I77a38c4611dc28db79d6f25d96ad0de36b224c98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d4a62f3348b36f4315d9e959910c4e4a4592fbfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d4a62f3348b36f4315d9e959910c4e4a4592fbfd"}]},"branch":"refs/heads/master"},"52ab68ec2e747aebfd91ebd2905f6b8f94a0bd76":{"kind":"REWORK","_number":2,"created":"2020-08-06 15:43:03.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/71/745171/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/71/745171/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/71/745171/2"}}},"commit":{"parents":[{"commit":"3f896a8d846de0682e2bfa5a8b45fc235620c7d5","subject":"Merge \"Remove neutron-fwaas from the jobs\u0027 required project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f896a8d846de0682e2bfa5a8b45fc235620c7d5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:33:40.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:42:41.000000000","tz":-420},"subject":"WIP: Begin work to modernize pbr\u0027s integration testing","message":"WIP: Begin work to modernize pbr\u0027s integration testing\n\nPBR does integration testing by installing all of openstack\u0027s python\nprojects to ensure we can install all of them in a variety of different\nways with different pip versions. Unfortunately, these jobs are still\nusing devstack-gate. We can simplify them significantly because zuul is\nnow doing what devstack-gate does and provides repos configured the way\nwe want them to be.\n\nThis change update the jobs to drop devstack-gate and simply consume\nzuul\u0027s repo prep.\n\nChange-Id: I77a38c4611dc28db79d6f25d96ad0de36b224c98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52ab68ec2e747aebfd91ebd2905f6b8f94a0bd76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52ab68ec2e747aebfd91ebd2905f6b8f94a0bd76"}]},"branch":"refs/heads/master"},"11407b554d16d91fcd757913d3903098b6c12f39":{"kind":"REWORK","_number":3,"created":"2020-08-06 16:03:13.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/71/745171/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/71/745171/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/71/745171/3"}}},"commit":{"parents":[{"commit":"3f896a8d846de0682e2bfa5a8b45fc235620c7d5","subject":"Merge \"Remove neutron-fwaas from the jobs\u0027 required project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f896a8d846de0682e2bfa5a8b45fc235620c7d5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:33:40.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 16:02:57.000000000","tz":-420},"subject":"WIP: Begin work to modernize pbr\u0027s integration testing","message":"WIP: Begin work to modernize pbr\u0027s integration testing\n\nPBR does integration testing by installing all of openstack\u0027s python\nprojects to ensure we can install all of them in a variety of different\nways with different pip versions. Unfortunately, these jobs are still\nusing devstack-gate. We can simplify them significantly because zuul is\nnow doing what devstack-gate does and provides repos configured the way\nwe want them to be.\n\nThis change update the jobs to drop devstack-gate and simply consume\nzuul\u0027s repo prep.\n\nChange-Id: I77a38c4611dc28db79d6f25d96ad0de36b224c98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/11407b554d16d91fcd757913d3903098b6c12f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/11407b554d16d91fcd757913d3903098b6c12f39"}]},"branch":"refs/heads/master"},"6fac9231fda2b0c19cd30c5e9938e223197c0e6a":{"kind":"REWORK","_number":4,"created":"2020-08-06 16:18:58.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/71/745171/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/71/745171/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/71/745171/4"}}},"commit":{"parents":[{"commit":"3f896a8d846de0682e2bfa5a8b45fc235620c7d5","subject":"Merge \"Remove neutron-fwaas from the jobs\u0027 required project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f896a8d846de0682e2bfa5a8b45fc235620c7d5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:33:40.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 16:18:51.000000000","tz":-420},"subject":"WIP: Begin work to modernize pbr\u0027s integration testing","message":"WIP: Begin work to modernize pbr\u0027s integration testing\n\nPBR does integration testing by installing all of openstack\u0027s python\nprojects to ensure we can install all of them in a variety of different\nways with different pip versions. Unfortunately, these jobs are still\nusing devstack-gate. We can simplify them significantly because zuul is\nnow doing what devstack-gate does and provides repos configured the way\nwe want them to be.\n\nThis change update the jobs to drop devstack-gate and simply consume\nzuul\u0027s repo prep.\n\nChange-Id: I77a38c4611dc28db79d6f25d96ad0de36b224c98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6fac9231fda2b0c19cd30c5e9938e223197c0e6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6fac9231fda2b0c19cd30c5e9938e223197c0e6a"}]},"branch":"refs/heads/master"},"cd2eaed5098d73e7fb3eab7a1493d43a483f4581":{"kind":"REWORK","_number":5,"created":"2020-08-06 16:58:48.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/71/745171/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/71/745171/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/71/745171/5"}}},"commit":{"parents":[{"commit":"3f896a8d846de0682e2bfa5a8b45fc235620c7d5","subject":"Merge \"Remove neutron-fwaas from the jobs\u0027 required project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f896a8d846de0682e2bfa5a8b45fc235620c7d5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:33:40.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 16:58:42.000000000","tz":-420},"subject":"WIP: Begin work to modernize pbr\u0027s integration testing","message":"WIP: Begin work to modernize pbr\u0027s integration testing\n\nPBR does integration testing by installing all of openstack\u0027s python\nprojects to ensure we can install all of them in a variety of different\nways with different pip versions. Unfortunately, these jobs are still\nusing devstack-gate. We can simplify them significantly because zuul is\nnow doing what devstack-gate does and provides repos configured the way\nwe want them to be.\n\nThis change update the jobs to drop devstack-gate and simply consume\nzuul\u0027s repo prep.\n\nChange-Id: I77a38c4611dc28db79d6f25d96ad0de36b224c98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd2eaed5098d73e7fb3eab7a1493d43a483f4581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd2eaed5098d73e7fb3eab7a1493d43a483f4581"}]},"branch":"refs/heads/master"},"9e229269b208671af5f4f3431b3f077dda5453c5":{"kind":"REWORK","_number":6,"created":"2020-08-06 17:46:11.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/71/745171/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/71/745171/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/71/745171/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/71/745171/6"}}},"commit":{"parents":[{"commit":"3f896a8d846de0682e2bfa5a8b45fc235620c7d5","subject":"Merge \"Remove neutron-fwaas from the jobs\u0027 required project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f896a8d846de0682e2bfa5a8b45fc235620c7d5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 15:33:40.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-08-06 17:41:27.000000000","tz":-420},"subject":"Begin work to modernize pbr\u0027s integration testing","message":"Begin work to modernize pbr\u0027s integration testing\n\nPBR does integration testing by installing all of openstack\u0027s python\nprojects to ensure we can install all of them in a variety of different\nways with different pip versions. Unfortunately, these jobs are still\nusing devstack-gate. We can simplify them significantly because zuul is\nnow doing what devstack-gate does and provides repos configured the way\nwe want them to be.\n\nThis change updates the jobs to drop devstack-gate and simply consume\nzuul\u0027s repo prep.\n\nA follow up will remove the old legacy jobs as they need to be removed\nfrom project-config first.\n\nNeeded-By: https://review.opendev.org/745189\nChange-Id: I77a38c4611dc28db79d6f25d96ad0de36b224c98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9e229269b208671af5f4f3431b3f077dda5453c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9e229269b208671af5f4f3431b3f077dda5453c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
