)]}'
{"id":"openstack%2Fpbr~780631","triplet_id":"openstack%2Fpbr~master~I7158e75934535b80eb60d6b4b7dcef08866b5bab","project":"openstack/pbr","branch":"master","topic":"fix-python_requires-support","hashtags":[],"change_id":"I7158e75934535b80eb60d6b4b7dcef08866b5bab","subject":"Revert \"Map requires-python to python-requires\"","status":"MERGED","created":"2021-03-15 17:29:42.000000000","updated":"2021-04-01 11:00:59.000000000","submitted":"2021-04-01 10:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780631-fix-python_requires-support","meta_rev_id":"7ca54bd7781b2c0bd7e02da890a42144d9475a3d","_number":780631,"virtual_id_number":780631,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 10:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-03-15 17:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-15 18:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-31 16:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 17:46:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-03-15 18:24:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-03-15 19:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 07:56:08.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"}],"messages":[{"id":"57744fc15d515b4fa0abb43446a44bce8152e2a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-15 17:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c86979dc7a3c7d840389caa39c1753bfe43efdf1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-03-15 17:46:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"57be32ca5e8c6189c3eafc55a1fcfe4b6802d89d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-15 18:24:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a6a2cc4beb3db2c575602653f1fccb374a02210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:06:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/04f65e10d53a44a18ead2366e85e3e1c : SUCCESS in 1h 29m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac8e2ce45fcb40418e2a716ef22fb706 : SUCCESS in 1h 25m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3095e047c34444ada3250c689efd00b9 : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199de72311ce48c48f031fd80353a1c6 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da118a821692490781488a9e66542ec1 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8b3507d6b724b9fbb8dd22d00e9ca9a : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4fbaebdafdf4f4da887c00fc0f1162c : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81200665511f44d4b978997e4fcfe502 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40965aee83f4abd863f40941c5c049f : SUCCESS in 5m 58s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/bb3c355e837841468504b5aa2f27011f : SUCCESS in 34m 33s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/2ad72bb5d8eb4a0f9847fc3b7ac4e91c : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3247bdc542e866185a5394a9810f3208acc93b24","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-03-15 19:34:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d39509bd7eea867fc7de00a7e556902a256ff62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 20:44:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d96b678388eb4af3b5374e891091932d : SUCCESS in 1h 08m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72bbec0dda984e5a97ab66ab9b61fe90 : FAILURE in 52m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9f49ca503b34331aafa779a7edba75b : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c155bf195f2747269635e7f3417eae70 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1513e60996b64c02b05244e2e4c908d3 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93255042691f48a98889922643fbb526 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afedf20dd3e744f39b5285713a16b467 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91f5f7be4b574f0c83f6d55421712e44 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e10c66c7c64038ac96a2f25f40c9c0 : SUCCESS in 3m 48s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a25ace362261449ab33114cb82cd0334 : SUCCESS in 50m 29s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/6e492a3b07314a88a46a17274f7e9b08 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2ee48c47ebbc0210f0932e74d0490270f2779448","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-18 11:01:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ce87a9ed9077e1baf2a977ff9c02ec76fe8e649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 13:24:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/944a21a68b4d44da8cef9e5691f73444 : SUCCESS in 1h 35m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d658a439209b4e9eb4673ae25f33f905 : SUCCESS in 1h 13m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd0ac75675604d4e87e86ff07b653841 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c275d740807456cbed3486647b48d62 : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/325baf473bc44b4b9706701146781d8a : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95a3dd034c274d90b9645654124052d1 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8a4c057da6347e1b712a1daa0143eeb : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2aa0300ade8d43b3947d204a20b1c234 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfba7dd33a484e8abce622ea752ebf2d : SUCCESS in 4m 55s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/29a282a4aa05431c9d25f0b5b7cfbb60 : SUCCESS in 34m 11s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/2a0f270cf2914ac09850c048fb969c27 : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6c6d343ea4a2fff61f09a5e03d13793bab8f8135","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-31 16:50:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a4a677b63326dfe5fcdedef1514add57c77d22d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:50:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d278d2d91d52dfa16c3c1dc70ffa1970b26c699e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:42:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a1b4f6a4c60418a8e782e7c28ef5194 : SUCCESS in 1h 29m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b95e65149194b0d96600112fb6190e7 : SUCCESS in 1h 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd195d07a6644ff9340c8b8bdd326f8 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4dcdfba2b8334956ab9df6794138f756 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2a81401d8d34f8197e9d13a544b6fd0 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b5487a9cc944178a3c206d4e674350f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca30af130454fc5999a5f9e60632132 : SUCCESS in 4m 47s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/73b5b01837a6436089ae8c4ca688b699 : FAILURE in 1h 15m 09s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/2c9b1bbf82bc408a8d7f075322d3c1df : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":1},{"id":"02a94315d9626a5922592cadd02d7f2899b1bea7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-31 22:09:25.000000000","message":"Patch Set 1:\n\nrecheck since it looks like the integration test failed somewhere in devstack","accounts_in_message":[],"_revision_number":1},{"id":"c787077b97e5b8902399d7cfc2b0ed20f07e3cb5","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-04-01 07:56:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd647d1d921431ee6b073fe1fa5afb6e0086a7c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 09:15:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/bf6c948ebf4f4c5e9421d9b3914ee35d : SUCCESS in 1h 14m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c56cf8d078d84a858111b010f1a7ae76 : SUCCESS in 1h 06m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d05aabad147d41e8bd5e3a6e63f9f988 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02cb1739901744479ce6ea7bc63cd749 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d41b3186a7646c4947e2d88f746cf1a : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbaf36d702f84d77a1eeb824b95d0b23 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1a997a4d0a0493bbbfd3273f83dd4ee : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/430b7e9142e54beb80dc51ab88201528 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ed93242fd44ceaba74e7f458cf3779 : SUCCESS in 3m 41s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/7ad7e133c5ec4e7ebe360f04e1652872 : SUCCESS in 25m 59s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/10bb4118cb6949a6b1b5b5525b4002be : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":1},{"id":"87475a4e878d97a85f59ae6c222c80573354f5f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 09:24:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f98d1ab2b42a09b1b2a2b48fe77f184c1ab2aedb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:59:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/62da0100ba364c45a38d77a1f96198ef : SUCCESS in 1h 31m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18607cd06b0449efb4464634413c940e : SUCCESS in 1h 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2177c75279842c4ba482a33666c3507 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b80c0e2f3744c20a41ee917a57e7db4 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95f03856ae1a4672969a1408d6c9e9b7 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6fb1e23f95949d5903dc660439112d4 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1fc7f839e0460da13f5e390f8d9a06 : SUCCESS in 3m 52s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/22e5f38fd8864ea18ece033b07e39d2d : SUCCESS in 29m 40s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/e1e15b0568ca4c819929c380137581cb : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b46ae3ae0887e87abf48713b4577829253bf5083","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:59:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ca54bd7781b2c0bd7e02da890a42144d9475a3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 11:00:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b728dd3a6a1a4fa9b38605dceb295aee : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5195593256f167d7843c14fa9b8b932bcd2afa34","revisions":{"5195593256f167d7843c14fa9b8b932bcd2afa34":{"kind":"REWORK","_number":1,"created":"2021-03-15 17:29:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/780631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/31/780631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/780631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/780631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/780631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/31/780631/1"}}},"commit":{"parents":[{"commit":"8e47cad9012c3781e82fc435499da4c1faca916e","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8e47cad9012c3781e82fc435499da4c1faca916e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-03-15 17:11:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-03-15 17:18:07.000000000","tz":0},"subject":"Revert \"Map requires-python to python-requires\"","message":"Revert \"Map requires-python to python-requires\"\n\nThis reverts commit d5da2121f875109a94809eeb8aa271fd5a0b3edd. This was\ndone incorrectly. The \u0027D1_D2_SETUP_ARGS\u0027 dictionary modified here is a\nmapping of distutils (and by extension, setuptools) supported options to\ntheir legacy distutils2 equivalents, and is used to transform the latter\nto the former. For example, consider the following entry:\n\n  \"long_description\": (\"metadata\", \"description\"),\n\nThis will transform an entry like so in a \u0027setup.cfg\u0027 file:\n\n  [metadata]\n  description \u003d ...\n\nInto the following equivalent call to setuptools\u0027 \u0027setup\u0027 function:\n\n  setup(\n      ...,\n      long_description\u003d\u0027...\u0027,\n      ...\n  )\n\nBy that measure, the entry added in this change, namely:\n\n  \"requires_python\": (\"metadata\", \"python_requires\"),\n\nWill result in a \u0027requires_python\u0027 argument being added to the \u0027setup\u0027\ncall. That\u0027s not what we want. Rather, we likely want to transform a\n\u0027[metadata] requires_python\u0027 entry to a \u0027python_requires\u0027 argument to\n\u0027setup\u0027.\n\nThe change is reverted here in expectation of a forthcoming proper fix.\n\nChange-Id: I7158e75934535b80eb60d6b4b7dcef08866b5bab\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5195593256f167d7843c14fa9b8b932bcd2afa34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5195593256f167d7843c14fa9b8b932bcd2afa34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
