)]}'
{"id":"openstack%2Fpbr~780632","triplet_id":"openstack%2Fpbr~master~Id43ba91585160448ad9737ba958ace5163d26f3d","project":"openstack/pbr","branch":"master","topic":"fix-python_requires-support","hashtags":[],"change_id":"Id43ba91585160448ad9737ba958ace5163d26f3d","subject":"util: Convert \u0027D1_D2_SETUP_ARGS\u0027 to a list of tuples","status":"MERGED","created":"2021-03-15 17:29:42.000000000","updated":"2021-04-01 12:31:54.000000000","submitted":"2021-04-01 12:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780632-fix-python_requires-support","meta_rev_id":"72d64f6dc78e460364cdeec4bd4fe17e208e2611","_number":780632,"virtual_id_number":780632,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 12:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-03-15 17:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2021-04-01 09:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-15 18:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-31 16:48:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-15 17:52:03.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-03-15 18:24:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-03-15 19:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 09:00:40.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"}],"messages":[{"id":"c7e551ec250fc6da71c4f270443558daae86f662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-15 17:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d5434822b22d5e8bc6dc0ca33bce4955a16dd5a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-03-15 17:52:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"916b9cd0950b1f75c3a110b026595577780b7df3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-15 18:24:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b50adab0a83a914b1cc7b4ff9561d0f49b725c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 19:07:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b7fe9d790fd415a97a8e38374ef70d6 : SUCCESS in 1h 30m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de26a59b3d964407ac74beb9e39ad75e : SUCCESS in 1h 30m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d14be67c6324605b5cb74902b332901 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92cd2890e16a4dada1c16530f47a44e7 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14a0b77eb2a34ffca24d3d20b56f6fc0 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1879c1d234404d48adf655c365b93e40 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b940c4117c224559a3c9d3571e38095e : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b43a6615f3394515996704da675932dc : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a90f66964154385b893beb01cb81833 : SUCCESS in 4m 59s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/172111274e1342a0919025a5d73bf547 : SUCCESS in 33m 58s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/017a88a686844d9d87e41d0ff6185c83 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"84c27abd9f03a2764cc63ba22222b4ada6abcfb0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-03-31 16:48:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a4423c932e4f7eac33a240f6bb4aa075d5991b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:50:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e01dd3da6a19ac691e0bbb8a3d035867d002015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:42:39.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"df2ff9d8c67d0d2debed4dfee359166d7f780df6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-01 08:59:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"24f3c3b9b31e22f43b6fdd34b915e0fdd0b65054","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-04-01 09:00:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"821756328987efd36080103a332bc7774c280a3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:48:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c7cbf1db5cd14279bb0d40ac8fb8c18d : SUCCESS in 1h 43m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8a25c52f32e4946aa04eaaa4a827f03 : SUCCESS in 1h 27m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/054b44548e32407095805b04a3b6eb46 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8966464153343abbd4ae675039d3ec4 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ca117ba36394585a2f8c98941ac08b2 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e68a71953aad4669a840b1df1f24f89d : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbb041eac61d4650874c5fc21dee684d : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffb97b6dd37c456ab368da2f6b9553bd : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d950d51c13c4f3dabb2dd2177757585 : SUCCESS in 3m 41s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/47a4aba6b5794ac2938dd454867bc25f : SUCCESS in 47m 52s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/85290e11dba24688a6432440a70c11ad : SUCCESS in 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2d35342acd8f452056d351b8427cf0b4d088ac9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:50:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80c76cbc7c230c8b4749794532a46bb28bcaaf3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 12:30:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/54f562433c2745849b8af6e8c978790c : SUCCESS in 1h 37m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d3a8f1ef314410e9f26ee5e325b955a : SUCCESS in 1h 36m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83c557d936047fa8323f32f20aa87a1 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cef66d0c49f444798158b74cba95c3d8 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a12e2c6e2b6d4a0998ea723fb4d18da2 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06bfcc07ad944dbb8cb0c0a3e8098d6f : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b8a60def914cbc886fbbe392ba0379 : SUCCESS in 4m 12s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/b254a410c84a45baa0878af3d3a83c89 : SUCCESS in 31m 37s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/344fb5171c6e4417ac185d347e3f47aa : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"16c0129a211b82887ae13bfa7d6f840bcf2112f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 12:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"72d64f6dc78e460364cdeec4bd4fe17e208e2611","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 12:31:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/975ce3c5481b417d8bfab9a6d9087293 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d6ae9c59a890a761c7d7b68d6d6cb5f33e1c867","revisions":{"1d6ae9c59a890a761c7d7b68d6d6cb5f33e1c867":{"kind":"REWORK","_number":1,"created":"2021-03-15 17:29:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/780632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/780632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/780632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/780632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/780632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/780632/1"}}},"commit":{"parents":[{"commit":"5195593256f167d7843c14fa9b8b932bcd2afa34","subject":"Revert \"Map requires-python to python-requires\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5195593256f167d7843c14fa9b8b932bcd2afa34"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-03-15 17:22:59.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-03-15 17:25:01.000000000","tz":0},"subject":"util: Convert \u0027D1_D2_SETUP_ARGS\u0027 to a list of tuples","message":"util: Convert \u0027D1_D2_SETUP_ARGS\u0027 to a list of tuples\n\nDicts can\u0027t have duplicate keys. Lists of tuples can. We\u0027re going to\nneed duplicates.\n\nChange-Id: Id43ba91585160448ad9737ba958ace5163d26f3d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1d6ae9c59a890a761c7d7b68d6d6cb5f33e1c867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1d6ae9c59a890a761c7d7b68d6d6cb5f33e1c867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
