)]}'
{"id":"openstack%2Fpbr~784931","triplet_id":"openstack%2Fpbr~master~I5007154caa7493652a424a4fad4918b7dbefcfb9","project":"openstack/pbr","branch":"master","topic":"py27","hashtags":[],"change_id":"I5007154caa7493652a424a4fad4918b7dbefcfb9","subject":"Update dependencies to avoid failure with old pip","status":"MERGED","created":"2021-04-06 11:50:14.000000000","updated":"2021-04-08 17:44:08.000000000","submitted":"2021-04-08 17:42:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784931-py27","meta_rev_id":"643d324511322186cc202c557d17433636b46310","_number":784931,"virtual_id_number":784931,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 17:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-04-07 14:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-08 15:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-08 15:52:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 11:50:45.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-04-06 11:50:45.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-04-06 14:04:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 08:24:27.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-04-08 15:52:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"0fd9a1b4cd3acd0c14a1d4d879458724dd704ed4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-04-06 11:50:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02cd4524a732ebe650834556d240eba47c47fe99","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-04-06 11:50:45.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"56db5fb5b7c8d02711ff0911381666046911deb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:04:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/22d7750291e34c5bb2b6c276c3a3aa2a : TIMED_OUT in 2h 06m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5afa424942e84309afb0a2af63dd2527 : TIMED_OUT in 2h 08m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2811595b1b7b4f72a105e7691b5506cd : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fd56499e2f4d079daed9a21abd9caa : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7465d6c284f44e7a092d80bd68e794f : FAILURE in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aba8e3cfbdf47b3acd3927cd079210b : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14e7c0d3b3534f9496c01a133842edf0 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f46f4ae073014c6d8686adf197ebaa45 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb708e3fc06421c9e9ab043c6966991 : SUCCESS in 4m 19s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/ac68963c179242ac8aebcd906b223aa8 : SUCCESS in 39m 39s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/f0b0a0c53ad44b45aa963bd824b8e46e : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b6c48e7951b59a6bc8f0b0e807e3675ac103207a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-04-06 16:43:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1012f8941c7cb7a371d8c47f8a2917eec7a486f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-04-06 16:44:25.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"277b7dac01ac22a80ddf3b530a710bd7fce3a7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:42:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/be54713824a145dca748edc1f86b8214 : SUCCESS in 1h 50m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12923786944a40128f27f3417dce7e2c : SUCCESS in 1h 24m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45c6d295a8ca4058b69d144738364025 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c30b9b65fc4d4a8980da6cbfbecf70 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6992c7db0ff64af181bc069b76dc8b94 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f617360816ae4fe0b7cb9e3c3a923f48 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5497e3151880485caa82885e6fb8d969 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d589af7f0f34c5ea7fc379041684617 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f16309f234c44c9285cf068082bf9aac : SUCCESS in 4m 25s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/713f675802fb4bd4a03a3656a0561e43 : SUCCESS in 40m 21s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/048548f291bc41d091a19f5eafa273da : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":3},{"id":"68114e348320398ed64918ac57f0b83fc50e5fce","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-04-07 14:48:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e8688e56a8fdc2382260d6c999156279064c8762","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-04-08 15:52:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"15eb97fc3f15a5688b1d9a8af7facd42c671eb45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 15:53:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"670d476bc7f3c73c431526424a11417acfe01469","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:42:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1e7ad1d4c19241e78f5c36cfc8fdbab8 : SUCCESS in 1h 47m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c332f3e398d946588bcbb3757de1667e : SUCCESS in 1h 24m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d1aa8bac344eeca2c8fd7545a25568 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd36b5e99417451691fb44f56a0ad5d5 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce8c1d2b976d40e48724adca1fdcf730 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a7964be50af45e3818c8e98a8b9a3b5 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3f26eef9804dae8591a6298e663678 : SUCCESS in 4m 26s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/b7f0ae0b3bfa4ec1817172e05155a0fc : SUCCESS in 34m 08s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/468b2cd974434bf2a5a59e610d0da22f : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":3},{"id":"0c21e41000b2baa74bff23bac65e823157a601a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:42:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"643d324511322186cc202c557d17433636b46310","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:44:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45905e588d3b4a8d8f494c245f44f993 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e0e2b8422a78b446c33ab3e5d1b38f8ad89b4bda","revisions":{"739e8e0d3ad7693be0465fe15b029bd952aab4fb":{"kind":"REWORK","_number":1,"created":"2021-04-06 11:50:14.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/31/784931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/31/784931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/31/784931/1"}}},"commit":{"parents":[{"commit":"2297988a7f41b8cb9d03aaea7d5ad9ba4378e54c","subject":"Merge \"Prevent test failure due to use of setpref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2297988a7f41b8cb9d03aaea7d5ad9ba4378e54c"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-04-06 11:49:59.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-04-06 11:49:59.000000000","tz":60},"subject":"Avoid py27 failure due to hacking","message":"Avoid py27 failure due to hacking\n\nWe do perform linting via pep8 job which runs on modern python only.\nThis fixes an error that recently broke py27 job due to failure to\ninstall hacking, where hacking is in fact not even needed by it.\n\nChange-Id: I5007154caa7493652a424a4fad4918b7dbefcfb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/739e8e0d3ad7693be0465fe15b029bd952aab4fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/739e8e0d3ad7693be0465fe15b029bd952aab4fb"}]},"branch":"refs/heads/master"},"22d8d6d696bc81bba6b82ffe090f88406b42d2c0":{"kind":"REWORK","_number":2,"created":"2021-04-06 16:43:03.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/31/784931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/31/784931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/31/784931/2"}}},"commit":{"parents":[{"commit":"2297988a7f41b8cb9d03aaea7d5ad9ba4378e54c","subject":"Merge \"Prevent test failure due to use of setpref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2297988a7f41b8cb9d03aaea7d5ad9ba4378e54c"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-04-06 11:49:59.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-04-06 16:42:27.000000000","tz":60},"subject":"Avoid py27 failure due to hacking","message":"Avoid py27 failure due to hacking\n\nWe do perform linting via pep8 job which runs on modern python only.\nThis fixes an error that recently broke py27 job due to failure to\ninstall hacking, where hacking is in fact not even needed by it.\n\nChange-Id: I5007154caa7493652a424a4fad4918b7dbefcfb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/22d8d6d696bc81bba6b82ffe090f88406b42d2c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/22d8d6d696bc81bba6b82ffe090f88406b42d2c0"}]},"branch":"refs/heads/master"},"e0e2b8422a78b446c33ab3e5d1b38f8ad89b4bda":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-04-06 16:44:25.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/31/784931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/31/784931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/784931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/31/784931/3"}}},"commit":{"parents":[{"commit":"2297988a7f41b8cb9d03aaea7d5ad9ba4378e54c","subject":"Merge \"Prevent test failure due to use of setpref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2297988a7f41b8cb9d03aaea7d5ad9ba4378e54c"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-04-06 11:49:59.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2021-04-06 16:43:14.000000000","tz":60},"subject":"Update dependencies to avoid failure with old pip","message":"Update dependencies to avoid failure with old pip\n\nWe do perform linting via pep8 job which runs on modern python only.\nThis fixes an error that recently broke py27 job due to failure to\ninstall hacking, where hacking is in fact not even needed by it.\n\nWe also need to add some extra conditions for mock to avoid possible\nfailure with older pip version.\n\nChange-Id: I5007154caa7493652a424a4fad4918b7dbefcfb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e0e2b8422a78b446c33ab3e5d1b38f8ad89b4bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e0e2b8422a78b446c33ab3e5d1b38f8ad89b4bda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
