)]}'
{"id":"openstack%2Fpbr~797898","triplet_id":"openstack%2Fpbr~master~I293f59b5074a38c78adffe580de2f1533bb01ce7","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2021-11-03 20:29:28.000000000","reason":"Change was submitted"},"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2021-11-03 20:29:28.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2021-11-03 20:29:28.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-11-03 20:29:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I293f59b5074a38c78adffe580de2f1533bb01ce7","subject":"Add a PEP517 interface","status":"MERGED","created":"2021-06-24 13:03:33.000000000","updated":"2021-11-03 20:31:30.000000000","submitted":"2021-11-03 20:29:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"797898","meta_rev_id":"4a56850ca816920c9e8e681b0e6e90b1ff21227f","_number":797898,"virtual_id_number":797898,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-03 20:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2021-11-03 17:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-03 17:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2021-11-03 17:35:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 13:51:16.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"CC"},{"updated":"2021-06-24 14:05:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-06-24 14:18:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-06-24 19:16:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 07:43:07.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REMOVED"},{"updated":"2021-10-26 19:55:46.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-10-27 07:15:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-10-27 07:15:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-11-03 16:51:31.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2021-11-03 17:29:45.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d3765e3aa56112260f1f47214b77b3b6b3f2c65d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2021-06-24 13:03:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eae47b5929bc313b7b290b5c9bd88e0670ec8e4d","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-06-24 13:51:16.000000000","message":"Patch Set 1:\n\npbr getting a 2021 facelist operation! Nice to see work towards PEP517.","accounts_in_message":[],"_revision_number":1},{"id":"ee129ede734e4792d482ec310067780a5f37f1fb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-06-24 14:05:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nI have no arguments against doing this, given it is the future of packaging in Python by all accounts. All that I\u0027d ask is that we document it as it would be nice to avoid more undocumented magic here 😄 I\u0027m happy to help out here if necessary. Also, I have a question inline about setuptools versioning plus some nits. The -1 is for the docs and setuptools versioning query.","accounts_in_message":[],"_revision_number":1},{"id":"a2ead3caa0cd94ff6e3a6c7f8d94a27fea48b9f8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 14:18:56.000000000","message":"Patch Set 1:\n\nNice changes.\n\nAll the hooks (mandatory and optional) are implemented.\n\nCould be worth to highlight a bit these changes with a release note.\n\nAlso maybe some unit tests could be added to that ensure things remain functional with time.","accounts_in_message":[],"_revision_number":1},{"id":"d2ce406077dca567509ccc020d528ba1269358db","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2021-06-24 15:28:33.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)\n\nHeya - yes - both to tests and docs. I put this together real quick this morning and wanted to get it up there. I\u0027ve also got some thoughts on next steps past this for improvements I\u0027ll write up.","accounts_in_message":[],"_revision_number":1},{"id":"d6b0f896d04c4cfa77e6046d3bfef2731b45d2f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:16:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e31721e859d84a539a702061fca96d4a : SUCCESS in 1h 30m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22bdde1c2afd4abdbad17f593faf499f : SUCCESS in 1h 33m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4d5402a748148c09f29a52fffaaec09 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c33003d93e74788baf7c931d870647c : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec010fc16d4b40368ca22eafcb1c2e07 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8a0179302a64465a86449a89ea40ee6 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09485a255eda4e79913f5297c38fec6a : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b9d9f0cc72a4a2cb9ba602d49c3da9b : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4045e242a0fb43c4a9e9c3cc3c40b19a : SUCCESS in 5m 47s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/abb7bcd508fa4a58b587f3a0dfb07e62 : SUCCESS in 41m 43s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/3720185cd15543f4aeb1cfd9f7fc9389 : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":1},{"id":"56ddfb42f3e9f7c83361e5975ab5240acba6d220","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2021-07-22 07:43:07.000000000","message":"Removed reviewer Sorin Sbârnea.","accounts_in_message":[],"_revision_number":1},{"id":"624e0a1b340267b51359d874d1088e7a0aa43266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-26 17:34:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2ad330776e6bf15642b7a63ae8202e27bfec6825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 19:14:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/062caa5c12184cc3a8ae8f55533048cb : SUCCESS in 1h 37m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/323aaf5338ca4e1cbb742d3007d47f2e : SUCCESS in 1h 29m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e19b7a3122d34365a1f91acd9dbcf1e9 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8927e0e43dd14c80a313a12f70a28244 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe1394349b0a4886a37ba5c904ae6735 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/609a5cdd401e4fd0a43f44660f10af1b : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/829fb5a4804641e2868e2e0b5c4707d7 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f537bd01609242ceb2f43d67b99c1596 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb610566104c464a82af5e9db6a5617a : SUCCESS in 4m 40s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/45712c837bd54ba2b870a577a0434ec9 : SUCCESS in 33m 02s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/4b5c33dfb4e04a7b86b23bc122d88d0d : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9f60976ea498ee7ad060594ef323fa6537b61e3e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-26 19:55:46.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"034917614144308e1c0e60e90735b6b28a28ad72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-26 23:04:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6eb083266741b6e2920a92c9b01d7429a3e2bf9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-26 23:06:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"830b7d28993584b70fdf53401c1c6e8846beefe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 00:46:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/18290542cfe041079940943a2d732b30 : SUCCESS in 1h 40m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e63d2883e2f47e382dfd651e1e195f8 : SUCCESS in 1h 17m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8eae2dd5ee4b519bbb30b74893819d : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ac16c454244c0b959c3b9292c8043a : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6d9ccf52df34d628a48fa67c19d10a6 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6ea7f1f848425da53b06965552fbf3 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e3156e12d234f4b961bddc5df96efbb : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7d9659faa414666a9d2fa50defb2578 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c357776b2f42f3a003d6fb51d782b7 : SUCCESS in 4m 20s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/56ced8ae57cb4d4087dd4b98d3511e36 : SUCCESS in 33m 12s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/71d2527758094125bce4d5882383eeef : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":3},{"id":"419e010db4eb6b5df8af471f2a273d8eeea647d1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-27 07:15:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b9133a18e7c3ceea86676d6fe56c0262b36857a8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-27 15:27:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ddeac399601a6eda30bb1feaac65da1ffc30bbfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-10-29 18:39:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52b49f4b19ff9abe4f5f87603fb35035780375e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 20:02:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1aa938ca019e4f458681ffea0617bfdc : SUCCESS in 1h 11m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d6344b3bc1f4a778cf998f0642e957b : SUCCESS in 1h 19m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6415b5a0746e4127aba6b4d41e9f6839 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6414c2da07242fcaad475bb573b36e7 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee8d719146d94263ad2131e6f5ea2890 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52fb0472ce974463ab3da1ea1018f8cb : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fd034f2324a4549af633ce4f23b8554 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/196d1c573afe407db6fe7129b61aab04 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18423b601eec46478cee17ec37882926 : SUCCESS in 5m 04s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/67837b6f54654cc48fd3423a02d98cf9 : SUCCESS in 35m 30s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/ea769e5f2d1e49e1ba19de8fcfb65aa3 : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":4},{"id":"f268b3f6833e05b758bde7ae39061875e7ee43dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-01 09:37:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab05e3aa65d390b3d68fbd6b5cf12e2dca5fdc92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-02 15:29:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a24d12c86cd13b17d72e92413dbdb68a3181a765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 17:01:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/957826e496ac445596851ebe0cf1a91e : SUCCESS in 1h 17m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adf947aec9674af194dffff64c3c3024 : SUCCESS in 1h 16m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ecd26314f6e4e068b6ff30682066ec3 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e55ca30ad0fb461ebe7eb8a777394b76 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad52abed79fe41ea86aa12d4e06cf9be : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc47824580e444a9c7ddc9c005ab5d2 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bdb9c2d7aaf49bb8a09467c5f3727fc : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9351b1dbb15e4397bbeb7a585232ac4d : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e789d5f01d4047ba9d94d33c1a1e6d9d : SUCCESS in 5m 34s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a8b244293f404560a15874f8132e37ac : SUCCESS in 31m 23s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/b2d4b6a5239d4636a1c3225c9ad33ed2 : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":5},{"id":"d597d710c0ce5dd20ee906c90516ae9e948cb8fb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-03 09:49:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0f9cdb0049c785578aa2e7d12bca8007110b1e7b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-03 16:51:31.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6ea30c7de39cd5da2d68bfbb599e475fb8201434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-03 17:06:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fac32e6536c553aa68c5efd80e9c7db52b79719f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-03 17:29:45.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2f0e7556136028e4f88a11a5cc53bf2ad214d994","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-03 17:35:38.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6577cc737e36a43a50f16c649156f30a4a889df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 18:51:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/161990fbce1547de9334248eb105e8d4 : SUCCESS in 1h 24m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/264e2738ee95480eac210d80ee276ad9 : SUCCESS in 1h 34m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40f6da902454b1da036bffd6cb3fbc4 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9b9618a9c747419d4c6cc3db4401a5 : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32c17f5442204db0a3bea150a93002c6 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d631cbd5b0864eae93a9f1d4e74ad353 : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43ba8bd86dda4fb5af09adfc61d7ca2f : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8c96bd790d64e3e8ecca36bbc767333 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/816a53df0257474b8e7644711ac17a57 : SUCCESS in 5m 49s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/19d6371a1b01492d8a8a17596d4fee54 : SUCCESS in 34m 58s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/bcd06177c55c40848c1339337284d6e5 : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":6},{"id":"54efa4b336ad5b26a69dfdc5a7a926b47af14e0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 18:55:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ae3cb9e6705a34c04dddf6238a15573c9651625c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 20:29:28.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c8834af34e294d3db751808f21af2392 : SUCCESS in 1h 32m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80c0ae4da6ee4be199641dfa0365de5f : SUCCESS in 1h 33m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28f95e328be4e098acbd00e465baf6f : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2609f602b4ee4cc2ba1dcc7de4ed1e5f : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6d69c4ed1e34f03a5a8826dea2f809e : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a9a654eda8941078ee065cc74801d48 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15bd744b845f415aa5a47a5d8e726c2e : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d036f5400b2f474ebc96a8f15ea7479c : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb8fea9c1fd74ffea998d36a911f1675 : SUCCESS in 5m 54s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/86bd3db0ee664828b4b23ea5306be1b7 : SUCCESS in 40m 07s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/ff1980b7a50b4094aace68d8127e603b : SUCCESS in 39m 51s","accounts_in_message":[],"_revision_number":6},{"id":"e85e9c1165f4fd5f5acb8f9676917c2ce6c3bdf2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 20:29:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"4a56850ca816920c9e8e681b0e6e90b1ff21227f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 20:31:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9faa50c49b42450c9b33b4a94143965c : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"09ee15341014fc0e3bb8a7c3b06a3fa912cfad38","revisions":{"a2ac416a181c2528187149a43d2c4bb1cbb03b15":{"kind":"REWORK","_number":1,"created":"2021-06-24 13:03:33.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/98/797898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/98/797898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/98/797898/1"}}},"commit":{"parents":[{"commit":"f3b1c7533e4fe2fdbe847f7471478e7953dc1b4f","subject":"Correct comment for \u0027D1_D2_SETUP_ARGS\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f3b1c7533e4fe2fdbe847f7471478e7953dc1b4f"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2021-06-24 12:24:30.000000000","tz":-240},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2021-06-24 13:00:51.000000000","tz":-240},"subject":"Add a PEP517 interface","message":"Add a PEP517 interface\n\npep517 defines a new module method of specifying build backends.\nTo allow pbr to exist in this world, we should define the interface\nthat\u0027s needed. For this to be used, one will put:\n\n  [build-system]\n  requires \u003d [\"pbr\"]\n  build-backend \u003d \"pbr.build\"\n\nInto pyproject.toml - and the pep517 interface will be used.\n\nThis doesn\u0027t really change anything else - it just makes us us this. So\nby itself this commit isn\u0027t SUPER helpful. But maybe let\u0027s take baby steps\nwith something this prone to strife, yeah?\n\nAfter this we can start teasing some things apart and doing our\nown things directly.\n\nChange-Id: I293f59b5074a38c78adffe580de2f1533bb01ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a2ac416a181c2528187149a43d2c4bb1cbb03b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a2ac416a181c2528187149a43d2c4bb1cbb03b15"}]},"branch":"refs/heads/master"},"46f8db0aa925f1bb0d1b6c7b8c9b89eb960cea6c":{"kind":"REWORK","_number":2,"created":"2021-10-26 17:34:44.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/797898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/98/797898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/98/797898/2"}}},"commit":{"parents":[{"commit":"4849d82da7e2f689206f980a6e609bc474fc6b48","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4849d82da7e2f689206f980a6e609bc474fc6b48"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2021-06-24 12:24:30.000000000","tz":-240},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-10-26 17:33:23.000000000","tz":-420},"subject":"Add a PEP517 interface","message":"Add a PEP517 interface\n\npep517 defines a new module method of specifying build backends.\nTo allow pbr to exist in this world, we should define the interface\nthat\u0027s needed. For this to be used, one will put:\n\n  [build-system]\n  requires \u003d [\"pbr\"]\n  build-backend \u003d \"pbr.build\"\n\nInto pyproject.toml - and the pep517 interface will be used.\n\nThis doesn\u0027t really change anything else - it just makes us support this.\nSo by itself this commit isn\u0027t SUPER helpful. But maybe let\u0027s take baby\nsteps with something this prone to strife, yeah?\n\nAfter this we can start teasing some things apart and doing our\nown things directly.\n\nChange-Id: I293f59b5074a38c78adffe580de2f1533bb01ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46f8db0aa925f1bb0d1b6c7b8c9b89eb960cea6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46f8db0aa925f1bb0d1b6c7b8c9b89eb960cea6c"}]},"branch":"refs/heads/master"},"e909de06f7197cce5c6fa6f5ad454feeda2e4c92":{"kind":"REWORK","_number":3,"created":"2021-10-26 23:04:25.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/797898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/98/797898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/98/797898/3"}}},"commit":{"parents":[{"commit":"4849d82da7e2f689206f980a6e609bc474fc6b48","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4849d82da7e2f689206f980a6e609bc474fc6b48"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2021-06-24 12:24:30.000000000","tz":-240},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-10-26 23:03:03.000000000","tz":-420},"subject":"Add a PEP517 interface","message":"Add a PEP517 interface\n\npep517 defines a new module method of specifying build backends.\nTo allow pbr to exist in this world, we should define the interface\nthat\u0027s needed. For this to be used, one will put:\n\n  [build-system]\n  requires \u003d [\"pbr\", \"setuptools\u003e\u003d36.6.0\", \"wheel\"]\n  build-backend \u003d \"pbr.build\"\n\nInto pyproject.toml - and the pep517 interface will be used.\n\nThis doesn\u0027t really change anything else - it just makes us support this.\nSo by itself this commit isn\u0027t SUPER helpful. But maybe let\u0027s take baby\nsteps with something this prone to strife, yeah?\n\nAfter this we can start teasing some things apart and doing our\nown things directly.\n\nChange-Id: I293f59b5074a38c78adffe580de2f1533bb01ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e909de06f7197cce5c6fa6f5ad454feeda2e4c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e909de06f7197cce5c6fa6f5ad454feeda2e4c92"}]},"branch":"refs/heads/master"},"7ad923a788de6537130955c1feb83e7e9634fd40":{"kind":"REWORK","_number":4,"created":"2021-10-29 18:39:59.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/797898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/98/797898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/98/797898/4"}}},"commit":{"parents":[{"commit":"4849d82da7e2f689206f980a6e609bc474fc6b48","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4849d82da7e2f689206f980a6e609bc474fc6b48"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2021-06-24 12:24:30.000000000","tz":-240},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-10-29 18:39:40.000000000","tz":-420},"subject":"Add a PEP517 interface","message":"Add a PEP517 interface\n\npep517 defines a new module method of specifying build backends.\nTo allow pbr to exist in this world, we should define the interface\nthat\u0027s needed. For this to be used, one will put:\n\n  [build-system]\n  requires \u003d [\"pbr\", \"setuptools\u003e\u003d36.6.0\", \"wheel\"]\n  build-backend \u003d \"pbr.build\"\n\nInto pyproject.toml - and the pep517 interface will be used.\n\nThis doesn\u0027t really change anything else - it just makes us support this.\nSo by itself this commit isn\u0027t SUPER helpful. But maybe let\u0027s take baby\nsteps with something this prone to strife, yeah?\n\nAfter this we can start teasing some things apart and doing our\nown things directly.\n\nChange-Id: I293f59b5074a38c78adffe580de2f1533bb01ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7ad923a788de6537130955c1feb83e7e9634fd40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7ad923a788de6537130955c1feb83e7e9634fd40"}]},"branch":"refs/heads/master"},"3384725d10c7828a9b175e449bb978e2d933e514":{"kind":"REWORK","_number":5,"created":"2021-11-02 15:29:56.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/797898/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/98/797898/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/98/797898/5"}}},"commit":{"parents":[{"commit":"4849d82da7e2f689206f980a6e609bc474fc6b48","subject":"Retire django-openstack-auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4849d82da7e2f689206f980a6e609bc474fc6b48"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2021-06-24 12:24:30.000000000","tz":-240},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-11-02 15:27:58.000000000","tz":-420},"subject":"Add a PEP517 interface","message":"Add a PEP517 interface\n\npep517 defines a new module method of specifying build backends.\nTo allow pbr to exist in this world, we should define the interface\nthat\u0027s needed. For this to be used, one will put:\n\n  [build-system]\n  requires \u003d [\"pbr\u003e\u003d5.7.0\", \"setuptools\u003e\u003d36.6.0\", \"wheel\"]\n  build-backend \u003d \"pbr.build\"\n\nInto pyproject.toml - and the pep517 interface will be used.\n\nThis doesn\u0027t really change anything else - it just makes us support this.\nSo by itself this commit isn\u0027t SUPER helpful. But maybe let\u0027s take baby\nsteps with something this prone to strife, yeah?\n\nAfter this we can start teasing some things apart and doing our\nown things directly.\n\nCo-Authored-By: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nChange-Id: I293f59b5074a38c78adffe580de2f1533bb01ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3384725d10c7828a9b175e449bb978e2d933e514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3384725d10c7828a9b175e449bb978e2d933e514"}]},"branch":"refs/heads/master"},"09ee15341014fc0e3bb8a7c3b06a3fa912cfad38":{"kind":"REWORK","_number":6,"created":"2021-11-03 17:06:07.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/98/797898/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/98/797898/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/98/797898/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/98/797898/6"}}},"commit":{"parents":[{"commit":"8c0d5c314108c09bbb3004d608a5a95bd81b8820","subject":"PBR package testing improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8c0d5c314108c09bbb3004d608a5a95bd81b8820"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2021-06-24 12:24:30.000000000","tz":-240},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-11-03 17:05:32.000000000","tz":-420},"subject":"Add a PEP517 interface","message":"Add a PEP517 interface\n\npep517 defines a new module method of specifying build backends.\nTo allow pbr to exist in this world, we should define the interface\nthat\u0027s needed. For this to be used, one will put:\n\n  [build-system]\n  requires \u003d [\"pbr\u003e\u003d5.7.0\", \"setuptools\u003e\u003d36.6.0\", \"wheel\"]\n  build-backend \u003d \"pbr.build\"\n\nInto pyproject.toml - and the pep517 interface will be used.\n\nThis doesn\u0027t really change anything else - it just makes us support this.\nSo by itself this commit isn\u0027t SUPER helpful. But maybe let\u0027s take baby\nsteps with something this prone to strife, yeah?\n\nAfter this we can start teasing some things apart and doing our\nown things directly.\n\nCo-Authored-By: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nChange-Id: I293f59b5074a38c78adffe580de2f1533bb01ce7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/09ee15341014fc0e3bb8a7c3b06a3fa912cfad38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/09ee15341014fc0e3bb8a7c3b06a3fa912cfad38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
