)]}'
{"id":"openstack%2Fpbr~817054","triplet_id":"openstack%2Fpbr~master~If4320da1e89a5c765e0bfe56b724364f9cba34f3","project":"openstack/pbr","branch":"master","topic":"pep517-clarification","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2021-11-11 21:11:58.000000000","reason":"Clark Boylan replied on the change"}},"hashtags":[],"change_id":"If4320da1e89a5c765e0bfe56b724364f9cba34f3","subject":"Clarify the need for setup.py with PEP517","status":"MERGED","created":"2021-11-08 16:57:31.000000000","updated":"2021-11-12 00:23:20.000000000","submitted":"2021-11-12 00:20:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817054-pep517-clarification","meta_rev_id":"b62478ad1b329eb7f020df2f6d5dfa1e532fc5c9","_number":817054,"virtual_id_number":817054,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 00:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-11-11 17:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-08 17:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-11 17:37:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 17:05:43.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-11-08 20:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-11 17:37:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b9aac1dd4eb730dfd6a2da853fbf9b62cfce33d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-08 16:57:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c1b2a5fd0fef168d1274875460de550e23b5b61","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-08 17:05:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"072eb888f8da6f1a1d24a137f6a8b55128ef6f73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 20:04:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/efdc6a2caeb14efea48f916e5ad23f0b : SUCCESS in 2h 04m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ed7ceaeb87b4d0b9bbc13a832afeaf5 : SUCCESS in 1h 32m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ec8a73ba53487eb873cd893ccb3cc6 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b897738ed5ec4902a94c4672a8dfbdb9 : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c35047db118541ec97b9cdc5e9b08891 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c05287e662d4f6aa8e679f444ba68cb : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e0f9fee1b5e497e98c85a3a12d991a9 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc658c80ff89481a83c02173104a6e1f : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93e61a5d2d7848f79fd39b331358db45 : SUCCESS in 5m 16s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/4cd86040dca64c18871c06a6f48c0c19 : SUCCESS in 31m 15s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/27b42c8dab064403818710ef93d23d03 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"97bfeea50bfe739ad868d88b29386fa109eaec4b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-11 17:37:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e5a78e83912b149f5edc9501c3295da02084868b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 17:39:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9551f358f15f7fd7d88cd24c2b2d5df553008b4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 19:07:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/14a2ca74e4da43b5b40da46a1622436c : SUCCESS in 1h 27m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21deae56bb014032a9d49301e9e9af7d : SUCCESS in 1h 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90693d8083084913bf91b58b3e994a3f : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09fa8654cd2c49d687d6b6c7ace60c98 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/402668930a4949579715cc6375135190 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24df60e2eeec4b3484013b1fee64b63a : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91c3cbbfe6b64daa8a7e42512838eb0e : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f19f88e7f194c2f9cc9a007f44b7b0d : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ae07dc6615c4b769d4357ca1ee36473 : SUCCESS in 4m 50s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/714bea5094ab427592b0510c4e2622a2 : SUCCESS in 37m 54s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/237891a16435414787b1544c77d8c352 : FAILURE in 41m 06s","accounts_in_message":[],"_revision_number":1},{"id":"aca7a8a43f3b0506750f9642710a881191c20dfb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-11 21:11:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aac32bc80d6a288a8d008c2006ea2ff92565af93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 22:36:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a5577f2a52394d09b57e7e13accf10db : SUCCESS in 1h 19m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25564e0fbe124707904b3ede09212f9e : SUCCESS in 1h 18m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3202b563c940a5a1fbf0b91b7b3d2a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53aabd869648469f81ea0ba5303b6e50 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4912b714e08a4012b75672a4bb87d538 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4525386876c4249820fa2ee60bd3acc : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1db612e0fb7847cfac8bb33ec155ba84 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baaec957ba384716b44b77e339018deb : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a88048708c164c0f9229b5477e8ecdf2 : SUCCESS in 6m 42s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/b1df8c6dcac84309a1145e9b86c8a750 : SUCCESS in 29m 32s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/fcd1c67906474db7ae4c4def589cf1cd : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b2110060c150f657b6d043c3c6a8a8875e4e1038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 22:38:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ed5e490d85d41e05b950a36acbcbe62ab494249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 00:20:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6ced2576848f44a19d6a672f0f1d6dd5 : SUCCESS in 1h 29m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6b8918b7407466787a325b3342cc82c : SUCCESS in 1h 36m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe12257c3df4a84beab43f291ecf313 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/311f4c9c0b1e461b84107a3589f49a6c : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9ab568f91754e16b868647a9a7850f2 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8c5f03f981f431abd439b67d37b8b18 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eacf3c05a008452a806bf1f54643c413 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ff7110fe76d443da30018872b7a3ddc : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b79f39ec56e4b05a48d0dc70e871ab9 : SUCCESS in 4m 58s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/f741907a7ad748adbd3cfb169475cee3 : SUCCESS in 34m 56s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/8d3011d3cd3d4415a5f9010bacfdecd9 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"966e2a60d094feccb2f6b1386f77fbf50a01124f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 00:20:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b62478ad1b329eb7f020df2f6d5dfa1e532fc5c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 00:23:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2cc7796f19e4b85ba40768a0a34c809 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d21c2caf1dbd69ba630075e48254131499f01afe","revisions":{"d21c2caf1dbd69ba630075e48254131499f01afe":{"kind":"REWORK","_number":1,"created":"2021-11-08 16:57:31.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/54/817054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/54/817054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/54/817054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/54/817054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/54/817054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/54/817054/1"}}},"commit":{"parents":[{"commit":"09ee15341014fc0e3bb8a7c3b06a3fa912cfad38","subject":"Add a PEP517 interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/09ee15341014fc0e3bb8a7c3b06a3fa912cfad38"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-11-08 16:56:18.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-11-08 16:56:18.000000000","tz":-480},"subject":"Clarify the need for setup.py with PEP517","message":"Clarify the need for setup.py with PEP517\n\nThere was a misunderstanding that PBRs PEP517 build hooks meant we no\nlonger needed a setup.py. But it turns out the setuptools build_meta\nhooks that we wrap call out to setup.py. This means PBR will continue to\nneed to be used in conjunction with a setup.py.\n\nThis is better for backward compatibiltiy anyway so not a huge loss.\n\nDocument this more properly and update the test to reflect it.\n\nChange-Id: If4320da1e89a5c765e0bfe56b724364f9cba34f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d21c2caf1dbd69ba630075e48254131499f01afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d21c2caf1dbd69ba630075e48254131499f01afe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
