)]}'
{"id":"openstack%2Fpbr~817794","triplet_id":"openstack%2Fpbr~master~I1f2b4d34e587389f7e11b99d000e14477cf5091b","project":"openstack/pbr","branch":"master","topic":"no-pep517-setup-requires","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2021-11-13 03:06:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f2b4d34e587389f7e11b99d000e14477cf5091b","subject":"Allow PEP517 without setup_requires","status":"MERGED","created":"2021-11-12 17:19:54.000000000","updated":"2021-11-13 03:10:46.000000000","submitted":"2021-11-13 03:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817794-no-pep517-setup-requires","meta_rev_id":"a6c90dc90363220878508d19490363176294195d","_number":817794,"virtual_id_number":817794,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-13 03:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-11-12 18:03:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-12 20:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-12 21:55:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 18:03:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-11-12 18:54:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-12 20:02:18.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"2357586275e088b25e71370e469a688b6632f5d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-12 17:19:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6262bd8e56af55e73970d59d35f93861636d1f05","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-12 18:03:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8278bc9a64ebc57ab735019895762e72f2e197a4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-12 18:03:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffcfb58c483724de6cb467b0fdba28291780b924","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 18:54:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1d01064b777c46aaaa32a8b3ffe98939 : SUCCESS in 1h 21m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/016cfde5890c4cc580a4801eef9b69f2 : SUCCESS in 1h 27m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fc1356224b4a4ca00d100feef64a7f : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef2ff59409a401bbd1d4b80b194b955 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7e4476da60b429ba172c132c575abd7 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/270d69df9b8c480ab19bdacdf856c5f9 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8b54ec0c4414b599337eb2d24635231 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f8753d502634d83a592746244f44442 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51e4c3e3c37f4decb44f8c61cda6b861 : SUCCESS in 5m 54s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/71f1963f894c405cbee89461f0813643 : SUCCESS in 48m 51s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/86c74f88afcc44079bc01ac46184b8e9 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"77cd4c3d9fb5286182517df01dc0fffde82db013","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-12 20:02:18.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027ll approve once I0ebaff67634857216ca8d565311ed7c63f687a16 merges so we make sure this isn\u0027t somehow breaking things on Python 2.7.","accounts_in_message":[],"_revision_number":1},{"id":"da9a47c13d8ddcfc61a7cbf3f384ec5f1fb7e347","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-12 21:55:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9616e5bc08d680b26b15e0d74aa8695c7cd6f5b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 21:57:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a109938118289e081d4641b40f2a634e23ba7f9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 23:41:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4f254249748c432f8ddcce2e5d887d05 : FAILURE in 1h 39m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d5b4a7aa7fc4ff88dfbf3b12650b74d : SUCCESS in 1h 25m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a242d40c43594f9cbe01175dfd859e04 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51698f6fb313452db47f91a4bbda6984 : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d51894d9887245bfa7eaacf48806b140 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0ae4d81983a4130855197c81368e957 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e9a6d90d167439da95b0b8cfba1515f : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29438b667b2742e6934ce225bf22b9cd : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f357db5856d1410d97b8744ddedc5c73 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d6468bbcc454f04a3c376b3d7e146ac : SUCCESS in 6m 06s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/246c8cadfcf246afbbb09fa731d355b8 : SUCCESS in 35m 29s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/9ac04e1f500146d1a1004eefede2072d : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":1},{"id":"11df4dd829e978dbcd31cd0f61ba64fed1fec1ff","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-11-12 23:45:16.000000000","message":"Patch Set 1:\n\nrecheck tempest failure","accounts_in_message":[],"_revision_number":1},{"id":"b150d6a9cd4e3486804bb5f27a6caa6e6e01b779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 01:26:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e75f5f23a614c00b0173898e0e18d3c : SUCCESS in 1h 21m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c28403714126481e9c727aa82c95eb9a : SUCCESS in 1h 32m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccca09d236b149e68f5bb57405052997 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d876b9b69c948adb525960b6f10b758 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de4629eaed2047d79f280375c8b6df27 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cde23722811545868a82fb27c8ff0ed8 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ebce7d214cc41a3b263445a577f8b9c : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f74de3d45774f72b365ab60bce0e44d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07fa0cfaf06e427c9eca6a9f408bf0f9 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5cdcb9b20e5463480106d789b81a81e : SUCCESS in 5m 09s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/57af1dddbdee4897b49adc8df24af234 : SUCCESS in 39m 15s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/041d269dcfa740e4be32bb0e09dbbcc7 : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d8052b3b9960f653d4cf1dce3e2dabb85dd430c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 01:28:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23a9571146e5a554cd040ef985b06c3d67f69465","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 03:06:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0c3386e87f5d46d5a3f38aa6a384069f : SUCCESS in 1h 38m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/789c941e570e4e8c9dbfed56965d2064 : SUCCESS in 1h 32m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fd50504cc04b62a680bbb0edab4e56 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b6fc3e008a48708e5e396f1030f332 : SUCCESS in 4m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7952ede01ad5437282d822707b677aea : SUCCESS in 7m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34cf9a2694244e148b4bb57813fcae73 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a6310f2a41b45338975099f5d8612e9 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c4610b2316e4be1b9d457d04f9bd6ab : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b87fc53b8b234589ae64b91ddfb8c888 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8ae53efd49d422d97f9eb52273b9ecc : SUCCESS in 5m 22s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/b934d14abca942a28a063905ef843278 : SUCCESS in 33m 39s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/a5542f34a73e419cadec15ea6074b21a : SUCCESS in 40m 24s","accounts_in_message":[],"_revision_number":1},{"id":"811de1b3606073b84be28a97606500338b6c0028","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 03:06:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6c90dc90363220878508d19490363176294195d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-13 03:10:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb451fcbf50417b9f160f176e9838aa : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de0c717b17841b0b14019130872f8f02a6aed217","revisions":{"de0c717b17841b0b14019130872f8f02a6aed217":{"kind":"REWORK","_number":1,"created":"2021-11-12 17:19:54.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/94/817794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/94/817794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/94/817794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/94/817794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/94/817794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/94/817794/1"}}},"commit":{"parents":[{"commit":"d21c2caf1dbd69ba630075e48254131499f01afe","subject":"Clarify the need for setup.py with PEP517","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d21c2caf1dbd69ba630075e48254131499f01afe"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-11-12 17:16:04.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2021-11-12 17:16:04.000000000","tz":-480},"subject":"Allow PEP517 without setup_requires","message":"Allow PEP517 without setup_requires\n\nWhen using PBR PEP517 support without setup_requires were were\ninfinitely recursing because Distribution.finalize_options calls pbr()\nand pbr() calls Distribution.finalize_options. With a setup_requires\nthis doesn\u0027t seem to happen because something is resetting dist.pbr\u0027s\nvalue from True (callable) to None (not callable) after the initial\npbr() call and before further recursion can occur.\n\nExplicitly enforce this behavior in pbr() via an explicit update of\ndist.pbr to None to avoid calling pbr() more than once resulting in the\nunwanted loop.\n\nChange-Id: I1f2b4d34e587389f7e11b99d000e14477cf5091b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/de0c717b17841b0b14019130872f8f02a6aed217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/de0c717b17841b0b14019130872f8f02a6aed217"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
