)]}'
{"id":"openstack%2Fpbr~825665","triplet_id":"openstack%2Fpbr~master~If372a33607fff884cd7bf888ee5a23a2fbb351b2","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-01-21 08:32:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If372a33607fff884cd7bf888ee5a23a2fbb351b2","subject":"Don\u0027t test with setuptools local distutils","status":"MERGED","created":"2022-01-20 18:51:06.000000000","updated":"2022-01-21 08:34:05.000000000","submitted":"2022-01-21 08:32:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825665","meta_rev_id":"f9ded349fe4f7f435d763c999d6512d1836ccf05","_number":825665,"virtual_id_number":825665,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-21 08:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-01-20 21:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2022-01-21 06:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-21 02:29:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2022-01-21 06:34:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 18:51:47.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-01-20 20:31:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 02:29:22.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-01-21 06:34:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"8f3a6f6bb7628b6363756c3786170c4f72726af5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-20 18:51:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3413773bb0f4ba4a0c38874effea5634b097cf61","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-20 18:51:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cad5667700fc00d00b3b3c4f9e1cf6ebabae0af5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-20 19:39:09.000000000","message":"Patch Set 1:\n\nClark has filed https://github.com/pypa/setuptools/issues/3038 which I\u0027m happy to update the commit message and TODO comment with if reviewers prefer.","accounts_in_message":[],"_revision_number":1},{"id":"5e5cab0a066f7b5805a0acc9e0f847a63348543c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 20:31:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e67c09c4372a4589b107d9fbb32be27a : SUCCESS in 1h 34m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dddc0af15ea64e8a8da8f96ff3f54e10 : SUCCESS in 1h 26m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184838698ba04fdda65b1a945e3b1929 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b8ee2c1d4548e68a770023458d65c0 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf4c9c3aed9c476ba65ef4a1cdab0e32 : FAILURE in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9c578480efd4168bfcb0f1689886474 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f730ca2de3b40c8886406340464fe37 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/707fd62ff2c24f7ca24c65d54a4671ca : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48e2277b2b844ff09a27c62a30d83f1c : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee466af940ac4b018494c6032a9c18b3 : SUCCESS in 4m 46s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a8eed8534fcd4790ab5e82225531e079 : SUCCESS in 37m 05s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/4840f908f1d443378346809de95b79db : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4eb8334a89a0b2760ce9ecc1f4f48dbbd2281246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-01-20 21:01:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d89d2c051905c23998d59211eca76a717f908cbc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-01-20 21:06:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bb9c52280bef7f86db36f7ef454dbb2a2ac7c06a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 22:39:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/aff97fc3d484411aaf675b3e72d7e285 : SUCCESS in 1h 30m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02216d1a60ec4db4ad62417bfa31a03e : SUCCESS in 1h 34m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce76ec1695341a693ec795e373df8e6 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352e2f743a6b4816aac06b91c24afeca : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/542f732003fc41e3b9a7d50191b7d15d : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82626a2a0bc04414932fbe814347251e : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f5c657f109d4ccc898939e175272fcc : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6596050bc65f4a1eae510ee33e9afe01 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/394d32673ef4416399f9f6f8a433de1b : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58bbb4c751f247edb261a23d57753a48 : SUCCESS in 4m 37s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a57e33e0cde6462993ff85bacd598655 : SUCCESS in 35m 19s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/35cfbaca5c934ca7bc84a86cfc4782b0 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":2},{"id":"fda27aca163dd5248b2ab2bd0b2f8f9cab208e45","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-01-21 02:29:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"152b73f9522dae37ac77713bdcee4c916d0aabf4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-01-21 06:34:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e4f7ff9b857514598664af4d5d3ba28f8702ff73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 06:36:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ca3dbc7c6c64af3169611f0d81384e9c3955e60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 08:32:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8c19a33cad2d4ae096a282aaa1eefeb6 : SUCCESS in 1h 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/893cc039d477403197d0bd3f10e3932e : SUCCESS in 1h 38m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f5348f2184449d9d296a9595b8d839 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f42bc32c5ac4c798b520c96b6f7cc3b : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b10fdd1edc694a288aa30130fcb81bab : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec00b2afd66844acb36dead7b119b75c : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2ba99fcebd847b2988e03167ae13ebd : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09f5bbaf50b042f9b66fcefe3811eb34 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4e6ca3518954a1aa00f55e940ee3c44 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee63217915674e2d963f21a2cc9ffffb : SUCCESS in 4m 05s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/95574921537647228c157a4316a7d6a8 : SUCCESS in 36m 18s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/c24470f62b81488dba627bc7b98682d7 : SUCCESS in 35m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c3451f78510738cc488b5dcca13f54e43d204fea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 08:32:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f9ded349fe4f7f435d763c999d6512d1836ccf05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 08:34:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512d0bd5d3ca47f2b10a30b46aef1c3f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c718425299af4f4df7b32adb90e8d183e101c6bc","revisions":{"686c93483deb5cb8f70c0edda2e34814078b430a":{"kind":"REWORK","_number":1,"created":"2022-01-20 18:51:06.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/65/825665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/65/825665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/825665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/825665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/825665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/65/825665/1"}}},"commit":{"parents":[{"commit":"b6b204e33eb396aaa3dc5576a062d0e5c0cc5d7e","subject":"Use context blocks for open() calls in packaging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b6b204e33eb396aaa3dc5576a062d0e5c0cc5d7e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-20 18:46:06.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-20 18:50:35.000000000","tz":0},"subject":"Don\u0027t test with setuptools local distutils","message":"Don\u0027t test with setuptools local distutils\n\nWe noticed some aberrations in log level/threshold handling as of\nSetuptools 60.0.0, and the only substantive change for that release\ntoggles the default selection for its internal (\"local\") distutils\nreplacement vs the \"stdlib\" one. For now, toggle the\nSETUPTOOLS_USE_DISTUTILS envvar in our testing back to the old\nvalue. This should be revisited once logging is handled better in\nSetuptools.\n\nChange-Id: If372a33607fff884cd7bf888ee5a23a2fbb351b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/686c93483deb5cb8f70c0edda2e34814078b430a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/686c93483deb5cb8f70c0edda2e34814078b430a"}]},"branch":"refs/heads/master"},"c718425299af4f4df7b32adb90e8d183e101c6bc":{"kind":"REWORK","_number":2,"created":"2022-01-20 21:01:52.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/65/825665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/65/825665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/825665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/825665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/825665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/65/825665/2"}}},"commit":{"parents":[{"commit":"b6b204e33eb396aaa3dc5576a062d0e5c0cc5d7e","subject":"Use context blocks for open() calls in packaging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b6b204e33eb396aaa3dc5576a062d0e5c0cc5d7e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-20 18:46:06.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-01-20 21:01:47.000000000","tz":0},"subject":"Don\u0027t test with setuptools local distutils","message":"Don\u0027t test with setuptools local distutils\n\nWe noticed some aberrations in log level/threshold handling as of\nSetuptools 60.0.0, and the only substantive change for that release\ntoggles the default selection for its internal (\"local\") distutils\nreplacement vs the \"stdlib\" one. For now, toggle the\nSETUPTOOLS_USE_DISTUTILS envvar in our testing back to the old\nvalue. This should be revisited once logging is handled better in\nSetuptools. https://github.com/pypa/setuptools/issues/3038\n\nChange-Id: If372a33607fff884cd7bf888ee5a23a2fbb351b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c718425299af4f4df7b32adb90e8d183e101c6bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c718425299af4f4df7b32adb90e8d183e101c6bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
