)]}'
{"id":"openstack%2Fpbr~827914","triplet_id":"openstack%2Fpbr~master~Ice845dc93925f516721b54d3269c8485a8278d75","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-02-04 23:39:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ice845dc93925f516721b54d3269c8485a8278d75","subject":"Correctly evaluate whether we\u0027re in a dist context","status":"ABANDONED","created":"2022-02-04 18:40:51.000000000","updated":"2022-02-04 23:39:29.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"77af0a0d1cdb39108b12a5767eabe1c6ad516150","_number":827914,"virtual_id_number":827914,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-04 21:14:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 18:42:24.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2022-02-04 19:39:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-02-04 21:14:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed9df5ccf647d7ab9f702c339f7aaa57e04cf50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 18:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf3e24993cb1b563d9ef8de8195ad005f82c44b8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-04 18:42:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00903fe6e43d74b07ef0fcfe757c83e70226a8b2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 19:03:35.000000000","message":"Patch Set 1:\n\nI did that in my initial local implementation, but realized it was redundant with how the existing code was already testing for presence, so was trying to stick with the same style. I\u0027m happy to initalize the global first and drop the hasattr LBYL if reviewers prefer it.","accounts_in_message":[],"_revision_number":1},{"id":"c474dc1e551a1352b2096cbb2fa479fe2814f514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 19:33:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b1fcdf5634418d2f3c4ac6ab7a22d7075c73a53","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-04 19:39:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dc4c9bc5828ea9dae4d9a66745415dbaf6c4e691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 20:06:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e89967abf81d8126a182a459bd337907c55a4f29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 20:50:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ceb0780805ae64b5846fe9e9dade2a122ceac70f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 21:14:24.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/095140f8b33344299b6a16399e146a84 : FAILURE in 13m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3d14c3d8469459b83f6f5684122018f : FAILURE in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e5999baeda84ef3b14c97d2aa0ed395 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8778a2a3cc524b88b1372b8a94526fec : FAILURE in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c757a2e3e08d41e38cb374c758febc76 : FAILURE in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6505526bb3dd4dc69974ee2676b70c7e : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de6c0841187548f9b63dbceecd04c453 : FAILURE in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b4a4fdccaaa499e842f23653bd3bd5c : FAILURE in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ed45c6b9d8d495ca9bd5aa174180342 : FAILURE in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42491bee91f14818901b0af6b838b264 : FAILURE in 5m 08s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/938be32edca7490f9f6e090de5c5a3f9 : FAILURE in 14m 59s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/14267c2448694ef4b9dbded3a65c5dbe : FAILURE in 13m 28s","accounts_in_message":[],"_revision_number":4},{"id":"77af0a0d1cdb39108b12a5767eabe1c6ad516150","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 23:39:29.000000000","message":"Abandoned\n\nGoing with alternative I407ae88ab8de4b61f94034b3d79a2ca7f7d79d16 instead.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"692d8a247f0e3cea995002fef7800b102cda4e8a","revisions":{"0722d8df9001a8b80b69c29cd28e3f19bc37a16c":{"kind":"REWORK","_number":1,"created":"2022-02-04 18:40:51.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/14/827914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/827914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/827914/1"}}},"commit":{"parents":[{"commit":"c04f5609046a4082bb0e2dfe01ac10c5b8fa9e46","subject":"Test that pbr.json is included in sdists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c04f5609046a4082bb0e2dfe01ac10c5b8fa9e46"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 18:37:34.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 18:37:34.000000000","tz":0},"subject":"Correctly evaluate whether we\u0027re in a dist context","message":"Correctly evaluate whether we\u0027re in a dist context\n\nWith the change to support PEP 517 build backend use cases without\nsetup_requires, a recursion-buster was added to toggle the\nsetuptools entrypoint off. Unfortunately, this happened before\nwriting the pbr.json file, and was how the write function identified\nwhether the call was within the context of a dist build.\n\nAdd a global module attribute when entering the setuptools\nentrypoint, so that it can be used as the new identifier when\nwriting metadata.\n\nChange-Id: Ice845dc93925f516721b54d3269c8485a8278d75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0722d8df9001a8b80b69c29cd28e3f19bc37a16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0722d8df9001a8b80b69c29cd28e3f19bc37a16c"}]},"branch":"refs/heads/master"},"b6cf2fa6da96932236fbb7728c0724447dd5cec2":{"kind":"REWORK","_number":2,"created":"2022-02-04 19:33:53.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/14/827914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/827914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/827914/2"}}},"commit":{"parents":[{"commit":"7e956f3f88f31425bfa4ad8327f8be359b207405","subject":"Test that pbr.json is included in sdists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7e956f3f88f31425bfa4ad8327f8be359b207405"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 18:37:34.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 19:33:23.000000000","tz":0},"subject":"Correctly evaluate whether we\u0027re in a dist context","message":"Correctly evaluate whether we\u0027re in a dist context\n\nWith the change to support PEP 517 build backend use cases without\nsetup_requires, a recursion-buster was added to toggle the\nsetuptools entrypoint off. Unfortunately, this happened before\nwriting the pbr.json file, and was how the write function identified\nwhether the call was within the context of a dist build.\n\nAdd a global module attribute when entering the setuptools\nentrypoint, so that it can be used as the new identifier when\nwriting metadata.\n\nChange-Id: Ice845dc93925f516721b54d3269c8485a8278d75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b6cf2fa6da96932236fbb7728c0724447dd5cec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b6cf2fa6da96932236fbb7728c0724447dd5cec2"}]},"branch":"refs/heads/master"},"4b7375d1d2636b3cc6da2184cdbe15124d062e9a":{"kind":"REWORK","_number":3,"created":"2022-02-04 20:06:58.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/14/827914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/827914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/827914/3"}}},"commit":{"parents":[{"commit":"7e956f3f88f31425bfa4ad8327f8be359b207405","subject":"Test that pbr.json is included in sdists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7e956f3f88f31425bfa4ad8327f8be359b207405"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 18:37:34.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 20:06:49.000000000","tz":0},"subject":"Correctly evaluate whether we\u0027re in a dist context","message":"Correctly evaluate whether we\u0027re in a dist context\n\nWith the change to support PEP 517 build backend use cases without\nsetup_requires, a recursion-buster was added to toggle the\nsetuptools entrypoint off. Unfortunately, this happened before\nwriting the pbr.json file, and was how the write function identified\nwhether the call was within the context of a dist build.\n\nAdd a global module attribute when entering the setuptools\nentrypoint, so that it can be used as the new identifier when\nwriting metadata.\n\nChange-Id: Ice845dc93925f516721b54d3269c8485a8278d75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4b7375d1d2636b3cc6da2184cdbe15124d062e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4b7375d1d2636b3cc6da2184cdbe15124d062e9a"}]},"branch":"refs/heads/master"},"692d8a247f0e3cea995002fef7800b102cda4e8a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-04 20:50:24.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/14/827914/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/827914/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/827914/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/827914/4"}}},"commit":{"parents":[{"commit":"91e41bc2ddc32c2e743fd66ad464036c47b670ea","subject":"Test that pbr.json is included in sdists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/91e41bc2ddc32c2e743fd66ad464036c47b670ea"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 18:37:34.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 20:50:17.000000000","tz":0},"subject":"Correctly evaluate whether we\u0027re in a dist context","message":"Correctly evaluate whether we\u0027re in a dist context\n\nWith the change to support PEP 517 build backend use cases without\nsetup_requires, a recursion-buster was added to toggle the\nsetuptools entrypoint off. Unfortunately, this happened before\nwriting the pbr.json file, and was how the write function identified\nwhether the call was within the context of a dist build.\n\nAdd a global module attribute when entering the setuptools\nentrypoint, so that it can be used as the new identifier when\nwriting metadata.\n\nChange-Id: Ice845dc93925f516721b54d3269c8485a8278d75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/692d8a247f0e3cea995002fef7800b102cda4e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/692d8a247f0e3cea995002fef7800b102cda4e8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
