)]}'
{"id":"openstack%2Fpbr~827922","triplet_id":"openstack%2Fpbr~master~I3398b2534e513039e922056e2cb3059768621e72","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-02-04 23:40:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3398b2534e513039e922056e2cb3059768621e72","subject":"Import pbr.util directly to break import loops","status":"ABANDONED","created":"2022-02-04 20:50:24.000000000","updated":"2022-02-04 23:40:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e0de0b8f6c76f5f5379ff7839356b40aee85614","_number":827922,"virtual_id_number":827922,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-04 21:19:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 21:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1194a7fafa1698bc5ac6f9ae9042cf54aac9faaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 20:50:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5d13bd06fd46b59b31b8979d63cef4b82712d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 21:19:27.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/576efce324164754b6a492f3e91df538 : FAILURE in 11m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c888811c6914557b8391faa88fabe8f : FAILURE in 19m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046d17f82269459eabaeca49128c20c3 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8fa54e31afc4b27bb8e44c73870c40b : FAILURE in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/963b54c723cf4a00a430018013f6236c : FAILURE in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6818702fb1e140e2b6f784c213eb08b5 : FAILURE in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c40e84a8725341bea84c49ed06f7f425 : FAILURE in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b03d5c54d1e4435a62f64450d9b36f8 : FAILURE in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1de1d6e7f914f598ace2e1a80721459 : FAILURE in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/546eeb404d5f438286a4ceed98d22311 : FAILURE in 4m 51s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/3f0d9af4e3114a7f99c39b16962cba03 : FAILURE in 13m 52s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/bf569c5f074043d8aa6bf1c7af103c67 : FAILURE in 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2e0de0b8f6c76f5f5379ff7839356b40aee85614","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 23:40:27.000000000","message":"Abandoned\n\nNo longer needed as a workaround with the I407ae88ab8de4b61f94034b3d79a2ca7f7d79d16 solution.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ab99d6d9e8f801058d86ce42cb3d77ce5c0a307","revisions":{"9ab99d6d9e8f801058d86ce42cb3d77ce5c0a307":{"kind":"REWORK","_number":1,"created":"2022-02-04 20:50:24.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/22/827922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/827922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/827922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/827922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/827922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/827922/1"}}},"commit":{"parents":[{"commit":"397b6fcdf953f7483805ff1501ec7f847fa372f3","subject":"remove explicit mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/397b6fcdf953f7483805ff1501ec7f847fa372f3"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 20:47:41.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 20:50:17.000000000","tz":0},"subject":"Import pbr.util directly to break import loops","message":"Import pbr.util directly to break import loops\n\nIf you have a cycle of \"from x import y\" across modules, Python\n2.7\u0027s importer gives up and raises ImportError. This does not occur\nif the module is imported canonically. In preparation for an\nupcoming change where we need to import core in pbr_json, switch\ncore\u0027s import of util to a loop-resistant style.\n\nChange-Id: I3398b2534e513039e922056e2cb3059768621e72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9ab99d6d9e8f801058d86ce42cb3d77ce5c0a307"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9ab99d6d9e8f801058d86ce42cb3d77ce5c0a307"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
