)]}'
{"id":"openstack%2Fpbr~827927","triplet_id":"openstack%2Fpbr~master~Ia2b79ae622ec4a0647984dc3be4266c85d352b11","project":"openstack/pbr","branch":"master","hashtags":[],"change_id":"Ia2b79ae622ec4a0647984dc3be4266c85d352b11","subject":"Always write pbr.json","status":"ABANDONED","created":"2022-02-04 21:38:27.000000000","updated":"2022-02-04 23:40:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f479a76737e21604d744f9297c3a1163710deb59","_number":827927,"virtual_id_number":827927,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-04 23:16:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 23:16:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e30e2656fb90b9972709f643a796ca70a75e6d42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 21:38:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de333f6e0f8da8b4f7084518fef40566c381ccce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 21:47:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a452635af200cd23b5b87ed35f833e59dd7a2933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:16:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/05f1406765a745a2a1127eed9b9e9366 : SUCCESS in 1h 23m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db45c34adc4043079aeada130b833fd4 : SUCCESS in 1h 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816379aebb1842279ae61859433f2f4a : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f16098c21cd406b98dd0d77600c6c42 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40535f8f40954083a39e78837a5139b4 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cea04d473a394474b14017e8684c79e2 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/936dd7da405c4d72b412eb92942a08b0 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34006b7e86854bb785099bc93f53eb49 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60699e994d35419584ab9373db0f469d : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f876fb7c8904f9d8d6020dcb7cbb582 : SUCCESS in 5m 02s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/6afda06e5f6346a3b5bb521bd5f3d768 : SUCCESS in 53m 52s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/0fcef3c385ca460696f50cf550702b65 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f479a76737e21604d744f9297c3a1163710deb59","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 23:40:03.000000000","message":"Abandoned\n\nWhile this seems to pass tests, I407ae88ab8de4b61f94034b3d79a2ca7f7d79d16 is probably the safer and more efficient option.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"acba296d32ad2121cbd10ec66d585b24f554d7cc","revisions":{"3f820f2d34f1cbb0316b0227293e7404c8ee3fda":{"kind":"REWORK","_number":1,"created":"2022-02-04 21:38:27.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/27/827927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/27/827927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/27/827927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/27/827927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/27/827927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/27/827927/1"}}},"commit":{"parents":[{"commit":"1a7417accf341b5e85a62738f7e74b7a584b1edf","subject":"Test that pbr.json is included in sdists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1a7417accf341b5e85a62738f7e74b7a584b1edf"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 21:35:46.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 21:35:46.000000000","tz":0},"subject":"Always write pbr.json","message":"Always write pbr.json\n\nDrop the safeguard which checks for PBR\u0027s SetupTools entrypoint\nbefore writing the pbr.json metadata file. If we\u0027re calling the\nwrite_pbr_json() function, it\u0027s safe to say we must want to create\nit regardless of whether the entrypoint is set. This becomes\nnecessary with the removal of setup_requires for PEP 517 builds,\nbecause we need to unset the entrypoint explicitly in order to avoid\nrecursion.\n\nChange-Id: Ia2b79ae622ec4a0647984dc3be4266c85d352b11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f820f2d34f1cbb0316b0227293e7404c8ee3fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f820f2d34f1cbb0316b0227293e7404c8ee3fda"}]},"branch":"refs/heads/master"},"acba296d32ad2121cbd10ec66d585b24f554d7cc":{"kind":"REWORK","_number":2,"created":"2022-02-04 21:47:28.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/27/827927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/27/827927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/27/827927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/27/827927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/27/827927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/27/827927/2"}}},"commit":{"parents":[{"commit":"1a7417accf341b5e85a62738f7e74b7a584b1edf","subject":"Test that pbr.json is included in sdists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1a7417accf341b5e85a62738f7e74b7a584b1edf"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 21:35:46.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 21:45:45.000000000","tz":0},"subject":"Always write pbr.json","message":"Always write pbr.json\n\nDrop the safeguard which checks that PBR\u0027s SetupTools entrypoint is\nTrue before writing the pbr.json metadata file. If the dist has a\n\"pbr\" attribute, it\u0027s safe to say this is a PBR-style dist. This\nbecomes necessary with the removal of setup_requires for PEP 517\nbuilds, because we need to toggle the entrypoint off after it\u0027s been\ncalled into once, in order to avoid recursion.\n\nChange-Id: Ia2b79ae622ec4a0647984dc3be4266c85d352b11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/acba296d32ad2121cbd10ec66d585b24f554d7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/acba296d32ad2121cbd10ec66d585b24f554d7cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
