)]}'
{"id":"openstack%2Fpbr~827931","triplet_id":"openstack%2Fpbr~master~I407ae88ab8de4b61f94034b3d79a2ca7f7d79d16","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-02-05 03:06:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I407ae88ab8de4b61f94034b3d79a2ca7f7d79d16","subject":"Avoid recursive calls into SetupTools entrypoint","status":"MERGED","created":"2022-02-04 21:58:41.000000000","updated":"2022-02-05 03:09:10.000000000","submitted":"2022-02-05 03:06:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"827931","meta_rev_id":"50caae4181b402f8ed3f4a0ed6764105e10601e3","_number":827931,"virtual_id_number":827931,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-05 03:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-02-04 23:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-02-04 23:39:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 22:02:03.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2022-02-04 23:28:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 23:39:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"6f5bc99033877fa1792745836c788e10d843eccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 21:58:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b686aacaa3e281122948bc4b146cdbeaa1464f7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-04 22:02:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83ece26ef7a0364f6abd6954ee0511c183b85b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:28:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/248565c8010e44bfb1253a67568a112b : SUCCESS in 1h 23m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5effcb50a92d48979f4a3207c5c3ab8e : SUCCESS in 1h 20m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293ccd0246e64079acb02a0f640cffc0 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da52c275d183443f8bc30b704aff1143 : SUCCESS in 4m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dddff7cee574173a9ec7c6ab8bf9638 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e72d03346514c108df817062a342142 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a010bd64d894bf6b805fdd3755ed50b : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0303f54a8444470f8449d024b9c3c8b9 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03e452ce35004508999cd4fe23399ceb : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44311c1afc804175b9367e7339f04992 : SUCCESS in 4m 01s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/5ab23c2165834eb58e85fff07416e385 : SUCCESS in 37m 12s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/d762ed4a9dd5492ca83cb063e529e746 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"273429ce5198242f0e04603d9559993159700feb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 23:36:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"822ff3f16d4071ce5b125f844018c67fb6a3d0fb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-02-04 23:39:02.000000000","message":"Patch Set 2:\n\nNote that the included test was exercised separately in Ic5e431aee3ed0447db2f838a853fa3c051f9889e in order to ensure it\u0027s able to correctly catch the regression we introduced.","accounts_in_message":[],"_revision_number":2},{"id":"290e4c9eba084a9913e67b00436e629347e0e39f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-02-04 23:39:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f4ca26730c145977ec52d688c3adc9e62fb592d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 01:21:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6978f2b240174eda87ee3e7d5cb889fc : SUCCESS in 1h 29m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30cae7b4832f4349ba41c30cc993c8ab : SUCCESS in 1h 37m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255ab285e54a4894bf8dcffcdb7766f1 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f8ea66499b4b57aee4e939e0dc3427 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edf0739643544904be698502f37fa450 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/813ee14345b54cac8a59838dda1ed5fb : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f93b3ddc43341daaafa78bda6bbcb85 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5197fd05e2a94da8bf7048ed50b85480 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b349e81094f1485bac3c0dde7d96a301 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c37552795ee4133b0046702d00c3dea : SUCCESS in 4m 28s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/1805d7c8439244fe80da95530c15f5d9 : SUCCESS in 38m 55s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/5b5a90b5237047589550a8b600317877 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":2},{"id":"bc3f3e585600c0b42895491e05c3496b2321a1f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 01:23:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57dbc7c04ce5a917404924f937bc23d9a836a957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 03:06:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb9daf1cdde8470b8a65c5e2aae12646 : SUCCESS in 1h 41m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dea466dd64d144e1a3727e6e03ef5fc2 : SUCCESS in 1h 34m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff35404cd44f44b7870576a913171542 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ae1391b06d44545b754c5ed55be70c9 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a11cbbeebecb495680bda6b015038072 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c5b6e8656ce5406293a04476d2821763 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b777bcfe91e845b891ea2d7807eb3b05 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ff0666a2e4d4301895374b55f463f45 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f43ddf7f3f484c91a8f775a008450643 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46b424b6cdd1436583ef494d8a1f0fb3 : SUCCESS in 4m 34s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/1ae19312232f4fd3b66cb4a6cf51a00d : SUCCESS in 34m 59s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/78b0fb5fc6ef4a179b22ffbbdb821666 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":2},{"id":"60a9f759f3f729aa4a0bccf6c84fb8e5d64e66f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 03:06:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"50caae4181b402f8ed3f4a0ed6764105e10601e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 03:09:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1e24b98abd4dcdbb024204156be993 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"949c47c97966b22bc3a7a33c08bc8535969e3fa0","revisions":{"4af97a846c0af986e220939f3087f05c28412b56":{"kind":"REWORK","_number":1,"created":"2022-02-04 21:58:41.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/31/827931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/31/827931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/827931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/827931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/827931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/31/827931/1"}}},"commit":{"parents":[{"commit":"1a7417accf341b5e85a62738f7e74b7a584b1edf","subject":"Test that pbr.json is included in sdists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1a7417accf341b5e85a62738f7e74b7a584b1edf"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 21:56:47.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 21:56:47.000000000","tz":0},"subject":"Avoid recursive calls into SetupTools entrypoint","message":"Avoid recursive calls into SetupTools entrypoint\n\nAdd a check in pbr.core.pbr() to make sure it hasn\u0027t already run,\nand return immediately if it has. This replaces the earlier attempt\nat loop-busting by setting the entrypoint to None in the dist, in\norder to support PEP 517 builds without setup_requires.\n\nChange-Id: I407ae88ab8de4b61f94034b3d79a2ca7f7d79d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4af97a846c0af986e220939f3087f05c28412b56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4af97a846c0af986e220939f3087f05c28412b56"}]},"branch":"refs/heads/master"},"949c47c97966b22bc3a7a33c08bc8535969e3fa0":{"kind":"REWORK","_number":2,"created":"2022-02-04 23:36:48.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/31/827931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/31/827931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/827931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/827931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/827931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/31/827931/2"}}},"commit":{"parents":[{"commit":"397b6fcdf953f7483805ff1501ec7f847fa372f3","subject":"remove explicit mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/397b6fcdf953f7483805ff1501ec7f847fa372f3"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 18:36:27.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-02-04 23:36:13.000000000","tz":0},"subject":"Avoid recursive calls into SetupTools entrypoint","message":"Avoid recursive calls into SetupTools entrypoint\n\nAdd a check in pbr.core.pbr() to make sure it hasn\u0027t already run,\nand return immediately if it has. This replaces the earlier attempt\nat loop-busting by setting the entrypoint to None in the dist, in\norder to support PEP 517 builds without setup_requires.\n\nAlso test that pbr.json is now included in the dist tree, in order\nto avoid future regressions.\n\nChange-Id: I407ae88ab8de4b61f94034b3d79a2ca7f7d79d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/949c47c97966b22bc3a7a33c08bc8535969e3fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/949c47c97966b22bc3a7a33c08bc8535969e3fa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
