)]}'
{"id":"openstack%2Fpbr~840633","triplet_id":"openstack%2Fpbr~master~Ib8e251ae9dd93a5eb356478d633f1fef271911d9","project":"openstack/pbr","branch":"master","topic":"drop-py36","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-11 14:24:45.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"Ib8e251ae9dd93a5eb356478d633f1fef271911d9","subject":"Adding python classifiers py38 \u0026 py39","status":"MERGED","created":"2022-05-05 10:06:10.000000000","updated":"2022-05-11 16:10:29.000000000","submitted":"2022-05-11 16:09:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840633-drop-py36","meta_rev_id":"be3a88b0023c7dd7d56504e2f76745b0d2c32f4c","_number":840633,"virtual_id_number":840633,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 16:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-05-05 17:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-05-11 14:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-05-05 22:59:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-05-11 14:24:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 11:44:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 17:07:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-05-05 22:59:40.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"fd17a18f43ab6e8fbe90e09464a948e4e48d97cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 10:06:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a2207f5c7153e8c5e12f4347a65737b17a68dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 10:06:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8290308c0fc492f8a61c68d5fd8c25494a9c7aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 11:44:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78f5c9f896744025a6bd5389a0d79564 : SUCCESS in 2m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/953e61e5bc0d4c5d8358aa9405471646 : SUCCESS in 1h 15m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/220ad0ee72944988a8b1068c820b7142 : SUCCESS in 1h 34m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c88ba57bc84f3785a7c0b08f1f0c01 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf26bf574ab64551a82a9526d9a6c929 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e73d06a993e40a380f2193a48f9f379 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10c014d185b44812a5dcc34e7c01e020 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d21364af3d284a3e80f996f2aae3b2e7 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4413855a14e34d67861caf6516f5c302 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69179407f45c4b09b1b71e8e4cd8008d : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a104c33760134565bc8113065750adcd : SUCCESS in 4m 19s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/7607274067894e118be6e2fe1cd0fe6c : SUCCESS in 40m 20s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/8d6fe5b0bd67475585ebc5fc5c0e77d4 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ee77b8c0768f6ffd80f399aff17474c48bf432c8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-05 17:07:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e611287af06325c2cdc68ad3f33cf4523e8ca50b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-05 22:59:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1806aba9b5201e37009d8c2590857f742523bf78","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-11 14:24:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cf18e27e4477238af70cc3c7328b456883209ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 14:27:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"27e060e1a8f101f5a34a10a688c4ee5822e62204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:09:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/20be8146ca8447bcaf677f512bc2d755 : SUCCESS in 2m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/67f11ed949b04d799708aada30ce3e49 : SUCCESS in 1h 37m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8407c86e552d4848bac3789687b5b317 : SUCCESS in 1h 34m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a73fd7262b44405b14905c436089f28 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5921a966e6e545c98f2270b8e1ec14ca : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d281dd8d226b4f55a634230f40f857fe : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fb0a35cd5d746f9bb1d112649ecfb22 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35acc7927c2443ac8db4a92adcbfdf5f : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d31a13b1e3804bb39c608e2c12e682f3 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/828622a5ec3943739cc869b3ec40072b : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/513064423d9e4bd8a136813319bc169f : SUCCESS in 4m 16s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/ea72c30df86a4ea4b8e9a843f05388a8 : SUCCESS in 44m 48s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/dc9b8e1f83ff43b987931e74585ebdcd : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0bcb27cce431bed052ac6c3ddb2aac8fb862978d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:09:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"be3a88b0023c7dd7d56504e2f76745b0d2c32f4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:10:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01cdf239125498ca14e58ac177829a4 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ded0ee1ff7236aa6bdc94cf5f4426b3816c98ef","revisions":{"77828d64fe8faa58a2d20a620e8151b037a728e8":{"kind":"REWORK","_number":1,"created":"2022-05-05 10:06:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/840633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/33/840633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/840633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/840633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/840633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/33/840633/1"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 10:05:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 10:05:41.000000000","tz":120},"subject":"Adding python classifier py38 \u0026 py39","message":"Adding python classifier py38 \u0026 py39\n\nChange-Id: Ib8e251ae9dd93a5eb356478d633f1fef271911d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/77828d64fe8faa58a2d20a620e8151b037a728e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/77828d64fe8faa58a2d20a620e8151b037a728e8"}]},"branch":"refs/heads/master"},"5ded0ee1ff7236aa6bdc94cf5f4426b3816c98ef":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-05 10:06:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/33/840633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/33/840633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/840633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/840633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/840633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/33/840633/2"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 10:05:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 10:06:18.000000000","tz":120},"subject":"Adding python classifiers py38 \u0026 py39","message":"Adding python classifiers py38 \u0026 py39\n\nChange-Id: Ib8e251ae9dd93a5eb356478d633f1fef271911d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5ded0ee1ff7236aa6bdc94cf5f4426b3816c98ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5ded0ee1ff7236aa6bdc94cf5f4426b3816c98ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
