)]}'
{"id":"openstack%2Fpbr~841043","triplet_id":"openstack%2Fpbr~master~Ie37ce3989abceec946e8b3b04be66af8c17044d3","project":"openstack/pbr","branch":"master","topic":"importlib_metadata_locked_lib","attention_set":{},"removed_from_attention_set":{"34828":{"account":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"last_update":"2022-05-13 10:05:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie37ce3989abceec946e8b3b04be66af8c17044d3","subject":"feat: check for specific importlib_metadata\u003e\u003d1.4","status":"ABANDONED","created":"2022-05-08 14:36:16.000000000","updated":"2022-05-13 10:05:00.000000000","total_comment_count":6,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"06fd10d55182a3a35b38a51cb6fb64278bd5ab1a","_number":841043,"virtual_id_number":841043,"owner":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-08 15:09:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"all":[{"value":-1,"date":"2022-05-09 11:56:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-05-09 13:41:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-08 14:36:16.000000000","updated_by":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"CC"},{"updated":"2022-05-08 15:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-09 11:56:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-05-09 11:56:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-05-09 13:41:10.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"}],"messages":[{"id":"b608bfac91e55e9ba112d0e01a2441ac031f4262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-08 14:36:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a653ed67efb87418f9ca22cf6c891ba9f7b42fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-08 15:09:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8afc774825cf4289983dd3fb1116af60 : FAILURE in 30m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf1959d19ecc4075b777997a3560e777 : FAILURE in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f5cc6e016743af85a7bc79df9a9939 : FAILURE in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb8d93ae0134159a8fecac7287f1434 : FAILURE in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37dc7578c860454a85acf26766e8a0d1 : FAILURE in 2m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e638e905f6c41f88016f1a4132a213b : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79fe197922e04390b1be3ad74f58f3ac : FAILURE in 2m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b74063ea37964d50bb900b55d5dfe2d4 : FAILURE in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c81e7414734f1bb555add6d324c2ea : FAILURE in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14b03fccb31c4bcabc51e61c0ee26a78 : FAILURE in 2m 53s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/fcdc2bced6b9488184cd82fbf8bdac2c : FAILURE in 10m 45s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/b85b4e1b5f1746c6a9675bbabc32748c : FAILURE in 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ba1c9163e001df983b7dc2030b2196c47610c287","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-09 11:56:50.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3e38821327623d7357e5a66c832c4f1e8de7a8ba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-09 11:56:58.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"3df85dcb877f0808fe2eda0b933b11c2842c4bad","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-05-09 13:41:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06fd10d55182a3a35b38a51cb6fb64278bd5ab1a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-13 10:05:00.000000000","message":"Abandoned\n\nDuplicate of https://review.opendev.org/c/openstack/pbr/+/841222","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd39a074ff6e40a8ec34ee75582a13e7fe27bfac","revisions":{"bd39a074ff6e40a8ec34ee75582a13e7fe27bfac":{"kind":"REWORK","_number":1,"created":"2022-05-08 14:36:16.000000000","uploader":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"ref":"refs/changes/43/841043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/43/841043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/43/841043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/43/841043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/43/841043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/43/841043/1"}}},"commit":{"parents":[{"commit":"e6ee869737357a140c258c69442fdd661f04660a","subject":"Merge \"Changed minversion in tox to 3.18.0.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e6ee869737357a140c258c69442fdd661f04660a"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-25 15:34:29.000000000","tz":60},"committer":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-08 14:36:07.000000000","tz":180},"subject":"feat: check for specific importlib_metadata\u003e\u003d1.4","message":"feat: check for specific importlib_metadata\u003e\u003d1.4\n\nChange-Id: Ie37ce3989abceec946e8b3b04be66af8c17044d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bd39a074ff6e40a8ec34ee75582a13e7fe27bfac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bd39a074ff6e40a8ec34ee75582a13e7fe27bfac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
