)]}'
{"id":"openstack%2Fpbr~841222","triplet_id":"openstack%2Fpbr~master~If67caac8d7ee7d5f22d0c6d262582b432d3370d1","project":"openstack/pbr","branch":"master","topic":"importlib_metadata_locked_lib2","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-05-13 10:03:57.000000000","reason":"stephenfin replied on the change"},"34828":{"account":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"last_update":"2022-10-13 11:04:24.000000000","reason":"Change was abandoned"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2022-10-13 11:04:24.000000000","reason":"Change was abandoned"},"4190":{"account":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"last_update":"2022-10-13 11:04:24.000000000","reason":"Change was abandoned"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-05-12 14:13:30.000000000","reason":"Removed by jpodivin by clicking the attention icon"}},"hashtags":[],"change_id":"If67caac8d7ee7d5f22d0c6d262582b432d3370d1","subject":"fix: check for the installed version of importlib_metadata. use it only if pbr supports it. Closes-Bug: #1972975 Change-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1","status":"ABANDONED","created":"2022-05-10 07:52:03.000000000","updated":"2022-10-13 11:04:24.000000000","total_comment_count":15,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"2843b22fa7f2416a6d97a0ab09cdac1d0f25f19d","_number":841222,"virtual_id_number":841222,"owner":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-29 14:54:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-10 08:25:02.000000000","updated_by":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-05-10 10:58:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-10 11:19:04.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"state":"REVIEWER"},{"updated":"2022-05-10 11:19:04.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-06-06 06:45:09.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REMOVED"},{"updated":"2022-08-24 09:39:18.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"}],"messages":[{"id":"c678970b50d349bacc2da4c4b4c47d73ea650187","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-10 07:52:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b988fd25e5f1563d4511d48847a2e1c0643e999","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-10 07:53:10.000000000","message":"Abandoned\n\nmoved to https://review.opendev.org/c/openstack/pbr/+/841222","accounts_in_message":[],"_revision_number":1},{"id":"7a2a439a916d0fda2e5994c79aa54576266d7d7d","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-10 08:25:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a97939fe7c9db62921be6b2427520c67c33068b","tag":"autogenerated:gerrit:restore","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-10 09:20:52.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"0f7a9f43bd0c524156454b0f1b6733546eeeac1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:58:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e46b2605de9540ee96ec6e407e4047a9 : SUCCESS in 1h 16m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7fcee3a6b5c43a3a3790d5f08ea83ef : SUCCESS in 1h 34m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef41717e606647e49cda82cb9db845bf : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a187293a7dd0444386b0e0f6f584adcf : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d109401c813445580508519091c2947 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/614261fd89c14f7e81f2915dbd521a30 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/293d0585abfe4909bb9d8ac0af099898 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea1e581ab3634e7f91303057ce8c965d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa5b62006ca4bc693ea40f6db918bf2 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5dab33feaa34d4f8e1f6a07a26fce50 : SUCCESS in 4m 19s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a7b927631e91404286fa66f621f87c4c : SUCCESS in 36m 26s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/55537e418709453895275badc40c20c2 : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"63b8157b22ac3fbf4e69f222e2209633ea31e954","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-05-10 11:19:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52b7755deee6a633f1085ad29547c6fd44c6c733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-11 08:51:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f5fdba6da484a4af6c8f6d5d7ba00b85b6e8b487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 10:40:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb3d825b41ee4f488af782df1dbd9f36 : SUCCESS in 1h 46m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b55d6e2252314eee81c78124d71ffcd4 : SUCCESS in 1h 28m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e906dd0774de419f88854d31e63428d5 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046c320525b14ad495d765d7b656ac2c : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dd2f91249b241458163a140686e7a0a : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c008386e3c0b4bce8877489c9da5ab42 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9712a70975af47f68867a823538d0ac8 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/140d257b6f344ac7907092ec185d387f : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9002081f25aa4bbebb9cf7d297836128 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/337644974d0244a19ebd61485ca28719 : SUCCESS in 3m 51s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/9d1181f596154474b0cb4f0405a5ae4f : SUCCESS in 40m 22s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/22726151250649efb78dd6a256c007a0 : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":2},{"id":"883ecbf08e39158277b82fac2db3ba3f9ad97b99","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-12 10:50:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbc18ebf14f64fa8c48a3f0b9746c81142a5e879","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-12 11:21:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f7c7c7ebcb691fd7bb1b26ebf6f221e6d58b0f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-12 11:52:16.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2becd807bd4510e5f004fa0ebe48ec6dccc34f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 13:16:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/e07dd5629bd647dd9ffd21e8ded65bb4 : SUCCESS in 1h 20m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59a47bd3a18d41d9811efe690032e036 : SUCCESS in 1h 21m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2bb6400cf74ebebea202bac231303a : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aeead9541d847b2bd21d83b179ba834 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/482e22658f3d435f803e2479df846aed : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac981a2f12eb48db8af8fcd8229ad7d1 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caf261dd04984031b9d1e130035e5277 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ae9c364ca9e492caf7455d0d134c8b3 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e8ab8b2b7cb4997bc1fd38a0de56b6a : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fadadb34567490e85289e9a84558345 : SUCCESS in 4m 17s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/7b13407a625c43ee95372d978ec8feae : SUCCESS in 39m 05s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/7231d9d63a7b4150a20c97624e487d08 : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":3},{"id":"a8c7b653ad707dbd3a89770639364a0b7f9bd58e","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-13 08:14:29.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1732c3cf3ea95d21d6ac7f486f9e2f4a9fd5ed2a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-13 10:03:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"14dcbc2da71e5190363028df8bf6e6d3a0863ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-15 11:18:57.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0806547adbab1e936fefedcc579fa866698a6422","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-15 11:22:12.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"3a1cc8110733b587b75c9da5d102779c30809ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"date":"2022-05-15 11:22:27.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"0e31b0124895d198fbc483f952f58abe2cd5b25e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-15 12:55:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9f82c8baa18c4f71a3f729ebb00d8a02 : SUCCESS in 1h 22m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c131ae2550db4433a38a5744bf567619 : SUCCESS in 1h 30m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e787611524b48ea8c61735549133658 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63cde415d35d48fc94a6c5928265a14d : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042f2497c98c4416801868a807c69cb6 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/566142fc649a4147b6723882fcdf4936 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fd3848bb3db45f4890340f5e3ba62ca : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59e8645d5ae8482b95fe51ebeebbecf1 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49c80cf5bfd04134ac3b5cf8c2da426c : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/990b78dd618b4060a12f0ee0441713fe : SUCCESS in 4m 22s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/014dfaa12f2a4d4683aa034c5704912d : SUCCESS in 36m 03s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/5385b10cb9cc4f5f860f91dde2761087 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":6},{"id":"8b7a0e6f8b623a231c2eab153566672fc52bfbba","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-06-06 06:45:09.000000000","message":"Removed reviewer Jiri Podivin.","accounts_in_message":[],"_revision_number":6},{"id":"513c2f739989acb0a223b628c698e345793f4ede","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-24 09:39:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3465df8f36699ac6ea8498f0cecb92ccc6b1666d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-29 13:11:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"efe162a03cf56c43f3668e04b11b5a3b478b7908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 14:54:13.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01a29d878381427cac7c97d892793c4c\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/adb7c24b4c0c447398b57117358cf61b : SUCCESS in 1h 39m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20a47531fcd34f2c869426e40287b028 : FAILURE in 19m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf577ebdb714656ac8a0c396212e929 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3262f90e374b05ae29b1105fe030b0 : FAILURE in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15e439fb84f6408f87a28b98e570268b : FAILURE in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6aa524a2d7c44558a86f5e360d12f3e8 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f398fd063337431d91d09aca9039013b : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/057823a7f0b74968b4b6d22fb77465fc : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9eff7cb6809c40c3870e2983cfe7639c : FAILURE in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00ac486195d44b4886ac5deb981d919c : FAILURE in 4m 53s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/226a992cc0a64937bc30d610cef9ceab : FAILURE in 12m 58s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/9250f01935084c20aceaec4e4fbce8b3 : FAILURE in 19m 38s","accounts_in_message":[],"_revision_number":7},{"id":"2843b22fa7f2416a6d97a0ab09cdac1d0f25f19d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-13 11:04:24.000000000","message":"Abandoned\n\nNo activity for a long time and no answer to my last question. I close it, feel free to reopen it if it was an error.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"da5cf31bd6a0d082c69a300dd8d7845537c4e28a","revisions":{"ddafac658174842530209c12adc76334301fda4f":{"kind":"REWORK","_number":1,"created":"2022-05-10 07:52:03.000000000","uploader":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"ref":"refs/changes/22/841222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/841222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/841222/1"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"committer":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"subject":"fix: importlib_metadata minimum requirements","message":"fix: importlib_metadata minimum requirements\n\nChange-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ddafac658174842530209c12adc76334301fda4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ddafac658174842530209c12adc76334301fda4f"}]},"branch":"refs/heads/master"},"b3a3f8f4108b6d71e71e102fef889a043b262a9b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-11 08:51:40.000000000","uploader":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"ref":"refs/changes/22/841222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/841222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/841222/2"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"committer":{"name":"Bar hochman","email":"ilnest.bar@gmail.com","date":"2022-05-11 08:51:40.000000000","tz":0},"subject":"fix: check for the installed version of importlib_metadata and use it only if pbr supports it. bug-report: https://bugs.launchpad.net/pbr/+bug/1972975 Change-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1","message":"fix: check for the installed version of importlib_metadata and use it only if pbr supports it.\nbug-report: https://bugs.launchpad.net/pbr/+bug/1972975\nChange-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b3a3f8f4108b6d71e71e102fef889a043b262a9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b3a3f8f4108b6d71e71e102fef889a043b262a9b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"41e72305c5b8297a6d733fbd044e3e751b405a05":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-05-12 11:52:16.000000000","uploader":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"ref":"refs/changes/22/841222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/841222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/841222/3"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"committer":{"name":"Bar hochman","email":"ilnest.bar@gmail.com","date":"2022-05-12 11:52:16.000000000","tz":0},"subject":"fix: check for the installed version of importlib_metadata. use it only if pbr supports it. bug-report: https://bugs.launchpad.net/pbr/+bug/1972975 Change-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1","message":"fix: check for the installed version of importlib_metadata.\nuse it only if pbr supports it.\nbug-report: https://bugs.launchpad.net/pbr/+bug/1972975\nChange-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/41e72305c5b8297a6d733fbd044e3e751b405a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/41e72305c5b8297a6d733fbd044e3e751b405a05"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0173848f212f90c7ac64108dff5bcf78a058dba9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-05-15 11:18:57.000000000","uploader":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"ref":"refs/changes/22/841222/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/841222/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/841222/4"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"committer":{"name":"Jochman","email":"ilnest.bar@gmail.com","date":"2022-05-15 11:18:57.000000000","tz":0},"subject":"fix: check for the installed version of importlib_metadata. use it only if pbr supports it. closes-bug: https://bugs.launchpad.net/pbr/+bug/1972975 Change-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1","message":"fix: check for the installed version of importlib_metadata.\nuse it only if pbr supports it.\ncloses-bug: https://bugs.launchpad.net/pbr/+bug/1972975\nChange-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0173848f212f90c7ac64108dff5bcf78a058dba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0173848f212f90c7ac64108dff5bcf78a058dba9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c449b16aa385bf6fa7edd867cf54cf1e04fa6b34":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-05-15 11:22:12.000000000","uploader":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"ref":"refs/changes/22/841222/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/841222/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/841222/5"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"committer":{"name":"Jochman","email":"ilnest.bar@gmail.com","date":"2022-05-15 11:22:12.000000000","tz":0},"subject":"fix: check for the installed version of importlib_metadata. use it only if pbr supports it. closes-bug: #1972975 Change-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1","message":"fix: check for the installed version of importlib_metadata.\nuse it only if pbr supports it.\ncloses-bug: #1972975\nChange-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c449b16aa385bf6fa7edd867cf54cf1e04fa6b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c449b16aa385bf6fa7edd867cf54cf1e04fa6b34"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3dc0125f5bb94f6c78af462b369ae4a8afe676de":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-05-15 11:22:27.000000000","uploader":{"_account_id":34828,"name":"Jochman","display_name":"Jochman","email":"ilnest.bar@gmail.com","username":"jochman"},"ref":"refs/changes/22/841222/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/841222/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/841222/6"}}},"commit":{"parents":[{"commit":"b0a1b7209d62c1e67c1e11dc96f98fd457d69352","subject":"Future-proofing pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0a1b7209d62c1e67c1e11dc96f98fd457d69352"}]}],"author":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"committer":{"name":"Jochman","email":"ilnest.bar@gmail.com","date":"2022-05-15 11:22:27.000000000","tz":0},"subject":"fix: check for the installed version of importlib_metadata. use it only if pbr supports it. Closes-Bug: #1972975 Change-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1","message":"fix: check for the installed version of importlib_metadata.\nuse it only if pbr supports it.\nCloses-Bug: #1972975\nChange-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3dc0125f5bb94f6c78af462b369ae4a8afe676de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3dc0125f5bb94f6c78af462b369ae4a8afe676de"}]},"branch":"refs/heads/master","description":"Edit commit message"},"da5cf31bd6a0d082c69a300dd8d7845537c4e28a":{"kind":"REWORK","_number":7,"created":"2022-08-29 13:11:54.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/22/841222/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/22/841222/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/22/841222/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/22/841222/7"}}},"commit":{"parents":[{"commit":"45d4c9f02fafb425e61a18bbd7a20ef456293a56","subject":"Merge \"Fix symbol identification in multiline message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/45d4c9f02fafb425e61a18bbd7a20ef456293a56"}]}],"author":{"name":"Bar Hochman","email":"ilnest.bar@gmail.com","date":"2022-05-10 07:48:01.000000000","tz":180},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-08-29 13:11:47.000000000","tz":120},"subject":"fix: check for the installed version of importlib_metadata. use it only if pbr supports it. Closes-Bug: #1972975 Change-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1","message":"fix: check for the installed version of importlib_metadata.\nuse it only if pbr supports it.\nCloses-Bug: #1972975\nChange-Id: If67caac8d7ee7d5f22d0c6d262582b432d3370d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/da5cf31bd6a0d082c69a300dd8d7845537c4e28a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/da5cf31bd6a0d082c69a300dd8d7845537c4e28a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
