)]}'
{"id":"openstack%2Fpbr~842410","triplet_id":"openstack%2Fpbr~master~I532a365aa0f1a3206edf6b01e821bd83ea04ab3c","project":"openstack/pbr","branch":"master","topic":"faster-startup","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-08-03 12:45:16.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-08-03 12:45:16.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-07-14 15:34:28.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I532a365aa0f1a3206edf6b01e821bd83ea04ab3c","subject":"Use stdlib importlib.metadata where possible","status":"MERGED","created":"2022-05-18 16:42:26.000000000","updated":"2022-08-03 12:46:32.000000000","submitted":"2022-08-03 12:45:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842410-faster-startup","meta_rev_id":"7fec6e4dfd47dab58a47dfdb10910b872e91b423","_number":842410,"virtual_id_number":842410,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-03 12:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-03 11:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":2,"date":"2022-07-14 15:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-03 11:10:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 18:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-14 14:50:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-07-14 14:50:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-08-03 11:10:47.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"8498861549a179df09a0ee87472a8e8ba73ef482","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-18 16:42:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7128c88cf55d74a70698e5df2cab57b20db0cb85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 18:20:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/15e89fd8d0424a86b6225eebabd4bc27 : SUCCESS in 1h 30m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd323e44dc184c3dbecc99c5b1337fbd : SUCCESS in 1h 33m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afccd3c824e84c88b743c72aa604697d : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b77222ce6c455c8da0bfd7bba27030 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/553257585b1f4916852847011ee24850 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1c51639b68b41f0a378b5fae1820d10 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45dcb4dacf83452bb941997d8a83e64f : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cf51cef37114b24bb2edd4e01cff674 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be262e949c714f719b8c70d412c95906 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eac720487c794e01b837506142cc7587 : SUCCESS in 4m 38s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/236180f93af84d92aa65c34079941334 : SUCCESS in 43m 03s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/c5a9c14645b14d6ca4b40155233e6c22 : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"00b52eeef59a2877115af30f80cce98f19c6ae4a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-14 15:34:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d81c3b74c4465d1cff219c5bd53ee4b95fa2e825","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-03 11:10:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0dcf49b50283e1ff3047fc99874c174c0127e522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 11:12:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d84fea16a7ecc475cc45998ccab41b211c64df0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:45:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b337fdf698e4d0e8eabac518c92c4ed\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0befe25b0e6428f804ea3ac81cd4e13 : SUCCESS in 1h 32m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/890f94b4ab8c49dbb4b84b0063102ce9 : SUCCESS in 1h 29m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918e967c605b42bf8a50b47bd18ba333 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c1ca8dc63242a283218b17a940419f : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be99b144ef2e4135a0c837c710e07510 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e245acaa3ffc42c78663ebc70d4f2947 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed6441f55808455186e317d4d9244ce1 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f96e84be6a9414da410b6e23beacdff : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad8021098bb646a29f9684cb43140113 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8141482e4bc450e8873e52245131f69 : SUCCESS in 4m 59s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/d2e060a50e284ff8b7ed06ca1048dc57 : SUCCESS in 42m 30s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/e9eed094123d493e95722e9d176ef3aa : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":1},{"id":"96ec287915c4e7f189d12be779a3a3f387961d71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:45:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7fec6e4dfd47dab58a47dfdb10910b872e91b423","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:46:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56f98832a95a4c3991acd3f069632db6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4e6b2524d44cc7bd2172722d2db450 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a6546a362e4a28839e642e9a8ee937191fe1d5e","revisions":{"6a6546a362e4a28839e642e9a8ee937191fe1d5e":{"kind":"REWORK","_number":1,"created":"2022-05-18 16:42:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/842410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/10/842410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/842410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/842410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/842410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/10/842410/1"}}},"commit":{"parents":[{"commit":"5ded0ee1ff7236aa6bdc94cf5f4426b3816c98ef","subject":"Adding python classifiers py38 \u0026 py39","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5ded0ee1ff7236aa6bdc94cf5f4426b3816c98ef"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-18 16:41:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-05-18 16:41:28.000000000","tz":60},"subject":"Use stdlib importlib.metadata where possible","message":"Use stdlib importlib.metadata where possible\n\nThe importlib_metadata package isn\u0027t guaranteed to be available, but the\nstlib version will always be there (at least on Python 3.8 or later).\n\nChange-Id: I532a365aa0f1a3206edf6b01e821bd83ea04ab3c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6a6546a362e4a28839e642e9a8ee937191fe1d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6a6546a362e4a28839e642e9a8ee937191fe1d5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}]}],"submit_requirements":[]}
