)]}'
{"id":"openstack%2Fpbr~845628","triplet_id":"openstack%2Fpbr~master~I7902c33a76c4cce8f50307843476e5d9435c9e2c","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"34946":{"account":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"last_update":"2022-06-15 07:28:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7902c33a76c4cce8f50307843476e5d9435c9e2c","subject":"Specify Changelog procedure","status":"MERGED","created":"2022-06-13 19:45:03.000000000","updated":"2022-06-15 07:29:23.000000000","submitted":"2022-06-15 07:28:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845628","meta_rev_id":"918dc5c9856d02f52246f185dee6de06806e3e4d","_number":845628,"virtual_id_number":845628,"owner":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-15 07:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-06-15 06:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-13 21:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-06-15 06:54:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-13 20:18:28.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-06-13 20:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 06:54:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"b1249c2568bd49c284b078be92a626ab53360fea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-06-13 19:45:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ada062d22a3a75387b85fab74da7c6cecad0132","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-13 20:18:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nLooks good, just one correction marked inline. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"914e39d1a4292a3ae2b4bd85217a10424d9dd83b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 20:49:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ae0976eda14af9949188817f7f4336 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da0942feee547a79ba8dc388eb8a59e : SUCCESS in 3m 26s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/4c190740e4624e24abf4ae031e66e62c : SUCCESS in 36m 26s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/d3bd77114a7d4f54b159239c91e3346b : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a8315d2346bf7f024241249386e32a99affc5738","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-06-13 21:03:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e10520baaf6cbe6b011eae38fa5456fc1029018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-06-13 21:07:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6c71d9ddbc76de1450ce405959440fc1ff241dc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-13 21:55:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aaee69672a546ea79acf22b5be15fdefd6827d68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 21:57:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfb4e84fcd04dcb91d2628ef73e675b : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f730fa7cea494e3ebc46861c35ef098c : SUCCESS in 3m 08s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/4506f9afda264ed5a7f26031afdbd676 : SUCCESS in 38m 18s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/bde80c562f514373b66b09f9dfc3adb6 : SUCCESS in 46m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e7e606a4ddd06156839e80124092002fec617598","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-15 06:54:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a66352736488af6e81497ffbffff18388322a461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 06:55:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86f8766ca918a36582494ad99ad0179536332559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 07:28:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1abcf8b0a3424bb20a1f590aab6ede : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b51e6546fa40eb8cab6b092912fc2e : SUCCESS in 3m 05s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a4d03f4990d84eada5964e231a69ef7a : SUCCESS in 30m 46s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/0a16ae166e9143ba887a6dfb4bb6798f : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4884924ae3af2d7e31c942043f262a2d1ddbe4e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 07:28:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"918dc5c9856d02f52246f185dee6de06806e3e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 07:29:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67d17017d2b4dc69175decf3d053525 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1eadf23b94846a0ebc8402e09190ed1de6dbffd0","revisions":{"26d50efe05269672f64d02244565e922b41455bf":{"kind":"REWORK","_number":1,"created":"2022-06-13 19:45:03.000000000","uploader":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"ref":"refs/changes/28/845628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/28/845628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/845628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/845628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/845628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/28/845628/1"}}},"commit":{"parents":[{"commit":"3793d87deb762f773eda1622ed87941c24bc253d","subject":"Allow leading spaces when determining symbols","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3793d87deb762f773eda1622ed87941c24bc253d"}]}],"author":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-13 19:39:31.000000000","tz":-240},"committer":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-13 19:43:42.000000000","tz":-240},"subject":"Specify Changelog procedure","message":"Specify Changelog procedure\n\nUpdate the CONTRIBUTING file to specify how the changelog is managed.\nThis allows contributors to follow best-practices to ensure the\nChangelog stays up-to-date with all changes.\n\nChange-Id: I7902c33a76c4cce8f50307843476e5d9435c9e2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/26d50efe05269672f64d02244565e922b41455bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/26d50efe05269672f64d02244565e922b41455bf"}]},"branch":"refs/heads/master"},"1eadf23b94846a0ebc8402e09190ed1de6dbffd0":{"kind":"REWORK","_number":2,"created":"2022-06-13 21:07:45.000000000","uploader":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"ref":"refs/changes/28/845628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/28/845628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/845628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/845628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/845628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/28/845628/2"}}},"commit":{"parents":[{"commit":"3793d87deb762f773eda1622ed87941c24bc253d","subject":"Allow leading spaces when determining symbols","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3793d87deb762f773eda1622ed87941c24bc253d"}]}],"author":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-13 19:39:31.000000000","tz":-240},"committer":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-13 21:06:43.000000000","tz":-240},"subject":"Specify Changelog procedure","message":"Specify Changelog procedure\n\nUpdate the CONTRIBUTING file to specify how release notes are managed.\nThis allows contributors to follow best-practices to ensure the\nauto-generated Changelog stays up-to-date.\n\nChange-Id: I7902c33a76c4cce8f50307843476e5d9435c9e2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eadf23b94846a0ebc8402e09190ed1de6dbffd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eadf23b94846a0ebc8402e09190ed1de6dbffd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
