)]}'
{"id":"openstack%2Fpbr~847841","triplet_id":"openstack%2Fpbr~master~Id799c2fd368d9df38a5c01fc2071283aecef67c6","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"34946":{"account":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"last_update":"2022-08-05 17:56:35.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-08-05 17:56:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id799c2fd368d9df38a5c01fc2071283aecef67c6","subject":"Fix symbol identification in multiline message","status":"MERGED","created":"2022-06-27 17:49:33.000000000","updated":"2022-08-05 17:57:38.000000000","submitted":"2022-08-05 17:56:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847841","meta_rev_id":"dcc547a817b9881527cb08341cf0ba90ddc57c7d","_number":847841,"virtual_id_number":847841,"owner":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-05 17:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-08-05 16:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-05 14:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-08-05 16:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-27 19:14:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 06:35:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-08-05 14:12:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-05 16:04:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"dcde9efbe0119789b597d2712e11850f3b031612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-06-27 17:49:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7765f5563e52088e1d93378c580e74fd5a891234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 19:14:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/67a5f2a0418b43ffbea4477dd931ecdb : SUCCESS in 1h 18m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8c08a0602c24884ab782761a04ce7e9 : SUCCESS in 1h 21m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b08f07ecd24d2ea7d1fae19d223a5c : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bed701a69d4444d95ac147a2194c91a : FAILURE in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a93d85878ca4697a81c4c20ba161de3 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3575f6d69834c9e9b7ed309d2ced1a8 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dedf3f96bbf94a05b05ab67de09b0949 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe6eddbd08554709bf21aa4e56ab0f0e : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d36c1de438c04094905d3ded5e2738c6 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8129765700404df38d39251f80fc5d7d : SUCCESS in 5m 24s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/6ea9a8a3fd2b411a88a1a2cf3517607c : SUCCESS in 39m 36s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/d11615ae7af847229badc2a4fe2d79da : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6942d3a5d835d070fbb61b912bc74eb65678e791","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-28 06:35:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4beeb30f7556157826d61b09d1a9b4ccfde1a832","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-06-28 12:41:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d00c52a644a322ae09743a3e0a385c02d17a13c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-08-05 13:55:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38cd64bce440026b4b5e18a1d2c109281be9b1a0","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-08-05 14:01:38.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d2c2fd15c536ac2b9c07f141b3e1bed7de2e7d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"date":"2022-08-05 14:08:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"91ae9566786c831621fa18418e159b8e365d431d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-05 14:12:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1a4cdeaaa16551641d3ee665e940ae1564c9b644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 16:01:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93798fc6cb504089b1214ebadae517f8\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1ac93b17eac544e892b644e3032c928f : SUCCESS in 1h 41m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73a525df39dc44669b377324087cabac : SUCCESS in 1h 50m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5e5d3a796b44cf8b8be332ff21cd06 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c659ff8f8248e79742dddded6e14b6 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0463f0ef1aa34f189ab6e2f0a7f29bea : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a20d3f000d04810bcfc01f07fef258f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b47854208b854ff198b40fae6f2f2f60 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d92bb1dc6ca4393a2ecd6f90c9a2bd7 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/113080213a024eb29ac225e19dda9496 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f93e27e335a429f92a3f74cc20cba8a : SUCCESS in 7m 42s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/d1c57318e93740c78aa38bd0b36c4216 : SUCCESS in 44m 07s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/de8f1b35fd8945159f2fa1343b675a26 : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":3},{"id":"a62f0c8f6e4e41d217b8887943710a78c4d88146","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-08-05 16:04:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c8422ccb2c09aed2b526f563141d8ee832d40a92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 16:06:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7abe181e9d8d625e0a67eca132303288445f63cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:56:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1e04d0ceb44f62bc211fee431d2de8\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/20bb3fb7649645618c222b2530294d7c : SUCCESS in 1h 40m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/570138c9ed5943e3b18eb42aeaec79b0 : SUCCESS in 1h 50m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e099391cf71f4a6292c72a7b6f408525 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a4f0a78a33433885c7a12becebf47c : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d34f1ad8acdb492d9547ae185e71dee4 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e003d21f7e4b41d794392c740515a1b4 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95300016f2ad4fc3be9babdfc319dd7e : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7214ee891ed6460bbb57efd6e5f1b65b : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/023bb3b906524880b92803fd563c474d : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3258be8ee7404cddb3766debdb2bc376 : SUCCESS in 4m 28s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/b89da93c50cf4b0bb18c640ab073219d : SUCCESS in 30m 59s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/0b8c33aec8e34888abecf8e74d196b3a : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":3},{"id":"8620bcb32daa724f8a82f139bb986dff3f4c8378","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:56:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dcc547a817b9881527cb08341cf0ba90ddc57c7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 17:57:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65edb4f446e74f19933796a4a494cd68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa64ee529cbe4a09b4216af7da062e07 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0aecbea81bcf8134bc76c17d4bc481a6f56e7bb2","revisions":{"303d03f6fb335bb3de75d6cea98beedaffada2c9":{"kind":"REWORK","_number":1,"created":"2022-06-27 17:49:33.000000000","uploader":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"ref":"refs/changes/41/847841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/41/847841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/41/847841/1"}}},"commit":{"parents":[{"commit":"1eadf23b94846a0ebc8402e09190ed1de6dbffd0","subject":"Specify Changelog procedure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eadf23b94846a0ebc8402e09190ed1de6dbffd0"}]}],"author":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-27 17:35:05.000000000","tz":-240},"committer":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-27 17:49:28.000000000","tz":-240},"subject":"Fix symbol identification in multiline message","message":"Fix symbol identification in multiline message\n\n* Fix an issue where a multiline message with leading spaces would cause\n  invalid symbol identification.\n* Add unit-tests around multiple symbols in a single line\n\nSem-Ver: bugfix\nChange-Id: Id799c2fd368d9df38a5c01fc2071283aecef67c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/303d03f6fb335bb3de75d6cea98beedaffada2c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/303d03f6fb335bb3de75d6cea98beedaffada2c9"}]},"branch":"refs/heads/master"},"936b6399c9376a14e024a9e15b5c9cf6d408e5a2":{"kind":"REWORK","_number":2,"created":"2022-08-05 13:55:59.000000000","uploader":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"ref":"refs/changes/41/847841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/41/847841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/41/847841/2"}}},"commit":{"parents":[{"commit":"1eadf23b94846a0ebc8402e09190ed1de6dbffd0","subject":"Specify Changelog procedure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eadf23b94846a0ebc8402e09190ed1de6dbffd0"}]}],"author":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-27 17:35:05.000000000","tz":-240},"committer":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-08-05 13:52:46.000000000","tz":-240},"subject":"Fix symbol identification in multiline message","message":"Fix symbol identification in multiline message\n\n* Fix an issue where a multiline message with leading spaces would cause\n  invalid symbol identification.\n* Add unit-tests around multiple symbols in a single line\n\nSem-Ver: bugfix\nChange-Id: Id799c2fd368d9df38a5c01fc2071283aecef67c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/936b6399c9376a14e024a9e15b5c9cf6d408e5a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/936b6399c9376a14e024a9e15b5c9cf6d408e5a2"}]},"branch":"refs/heads/master"},"0aecbea81bcf8134bc76c17d4bc481a6f56e7bb2":{"kind":"REWORK","_number":3,"created":"2022-08-05 14:08:26.000000000","uploader":{"_account_id":34946,"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","username":"chris.dohmen"},"ref":"refs/changes/41/847841/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/41/847841/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/41/847841/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/41/847841/3"}}},"commit":{"parents":[{"commit":"1eadf23b94846a0ebc8402e09190ed1de6dbffd0","subject":"Specify Changelog procedure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eadf23b94846a0ebc8402e09190ed1de6dbffd0"}]}],"author":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-06-27 17:35:05.000000000","tz":-240},"committer":{"name":"Chris Dohmen","email":"chris.dohmen@sciencelogic.com","date":"2022-08-05 14:08:09.000000000","tz":-240},"subject":"Fix symbol identification in multiline message","message":"Fix symbol identification in multiline message\n\n* Fix an issue where a multiline message with leading spaces would cause\n  invalid symbol identification.\n* Add unit-tests around multiple symbols in a single line\n\nSem-Ver: bugfix\nChange-Id: Id799c2fd368d9df38a5c01fc2071283aecef67c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0aecbea81bcf8134bc76c17d4bc481a6f56e7bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0aecbea81bcf8134bc76c17d4bc481a6f56e7bb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[]}
