)]}'
{"id":"openstack%2Fpbr~884938","triplet_id":"openstack%2Fpbr~master~If3594269495ae1f3cefffd4c4f38d87af5c7fdfe","project":"openstack/pbr","branch":"master","topic":"cleanup-sphinx-integration","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-06-01 15:29:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3594269495ae1f3cefffd4c4f38d87af5c7fdfe","subject":"Use _ instead of - in setup.cfg directives","status":"MERGED","created":"2023-05-31 20:26:02.000000000","updated":"2023-06-01 15:30:37.000000000","submitted":"2023-06-01 15:29:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884938-cleanup-sphinx-integration","meta_rev_id":"2df3801b49e522a8b3fe3855f0b6980792542fe3","_number":884938,"virtual_id_number":884938,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 15:29:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-06-01 09:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-06-01 13:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-06-01 13:26:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-31 21:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 09:07:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-06-01 13:26:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"47e9fdc1779ccd004d56bbe1f380c3eaf6477f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-31 20:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8d02a8d2c8978a124fefb6842065efd5694af35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 21:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b276d82975a42a69457b64e2f67a48a\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff32546869e2419d9941798eed4d5457 : SUCCESS in 1h 17m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/071f242a4e3d4f6b8fa7d8c4c2f3791b : SUCCESS in 1h 19m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40dd68272e1746ada8778b5aa25e71fd : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19c2a6fd0c4407da146f8b0492beaa1 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daf1ab990c7e4bd0b86c72476863dba0 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dd089f35b1d467c8a03b1170043fbe0 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9119eacf99b47de8dd23bbb10c026ce : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e76723c680f40e59f6ce9e8b49fa8b7 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8e7aa2353884b658011a14bc20bdf8d : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d86c15dbcc648078b584b2d0982acac : SUCCESS in 3m 41s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/34eb753970954ad7901ed8cc700a59d7 : SUCCESS in 30m 52s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/e7b4cee29e3c4798bc17c9c9f0630020 : SUCCESS in 40m 11s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/11ec0bb2060948a7966d13bf50a04de1 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"152e03cf3b2efee5c12e111648fc5e51cb968e4b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-01 09:07:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2ebd997d42db5c3320c63df5c149a994ebe6f66","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-06-01 13:26:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af2b489421ffc0a1fcd74a744221eb548fb4585d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 13:28:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a8b85f56d3b7f2c10e3e00de30413810dbf73f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 15:29:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1fdf1389fb44610a7c2bd5386a6f15c\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c38f7cbc28fc4f9b8a16b8d37eb26ec5 : SUCCESS in 1h 53m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f4a22e89375463ebc4997c4329e581f : SUCCESS in 1h 34m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81683ac885144f87a92adc50474d459a : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b152d7029d49929d71902d407fbd14 : SUCCESS in 3m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b1e9ad716924fafb094536433fe56a3 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c79beabbeb35416aa970d039cc3f6302 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7579c7c2d8774c42b4f882d633bb1275 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7ba33d47c73477d969d8792377eeae9 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75989f702afd40b99c65a87f4244c7af : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a56d008341ee4d56be172a22be553ed5 : SUCCESS in 5m 20s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/5907ab95da2844bc9a384aec50fbec9e : SUCCESS in 35m 02s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/1c7deccb52044636b1f81f9859c2f2bd : SUCCESS in 32m 07s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/94156ade3c644d17b450fac350350c8b : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0850a479fe990d74bf3f577d25743c3e8c39a5b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 15:29:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2df3801b49e522a8b3fe3855f0b6980792542fe3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 15:30:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/565b5504806f4983a4a4e36044c63cc5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ed5900f7f8f4eeba0508871f19a1a96 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee04b62de28061018496d6324639446afe2059c7","revisions":{"ee04b62de28061018496d6324639446afe2059c7":{"kind":"REWORK","_number":1,"created":"2023-05-31 20:26:02.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/38/884938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/38/884938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/38/884938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/38/884938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/38/884938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/38/884938/1"}}},"commit":{"parents":[{"commit":"47c5afe79aa334b300b9f15f3f23ff76d812d742","subject":"Remove sphinx doc building integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/47c5afe79aa334b300b9f15f3f23ff76d812d742"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-05-31 20:25:23.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-05-31 20:25:23.000000000","tz":-420},"subject":"Use _ instead of - in setup.cfg directives","message":"Use _ instead of - in setup.cfg directives\n\nSetuptools deprecated -\u0027s at some point and they result in a bunch of\nwarnings in failed tests. This is distracting and eventually is likely\nto create real errors. Go ahead and fix them now.\n\nChange-Id: If3594269495ae1f3cefffd4c4f38d87af5c7fdfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ee04b62de28061018496d6324639446afe2059c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ee04b62de28061018496d6324639446afe2059c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
