)]}'
{"id":"openstack%2Fpbr~885064","triplet_id":"openstack%2Fpbr~master~I97bdda4ce89b26c6e99de6f1d6ad7d734cac9041","project":"openstack/pbr","branch":"master","topic":"build_sphinx_releasenote","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-06-01 17:38:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I97bdda4ce89b26c6e99de6f1d6ad7d734cac9041","subject":"Add release note for build_sphinx removal","status":"MERGED","created":"2023-06-01 15:25:54.000000000","updated":"2023-06-01 17:39:26.000000000","submitted":"2023-06-01 17:38:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885064-build_sphinx_releasenote","meta_rev_id":"a2e71b0b74dd2665227a5ca030b8449b3c5d95d6","_number":885064,"virtual_id_number":885064,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 17:38:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-01 15:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-01 15:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 15:30:45.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-06-01 16:40:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b12927d05530b29e80243219b26f913130eae22f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-06-01 15:25:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fb2ee5e1337048d451c31b1a321e2867fe4774f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-06-01 15:30:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28163b15f66d5084497fd3e88e5e9fc8c050a461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 16:40:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5051fbd98934b5fa667f228c9b45a27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1d9ab030d5428a9e39e34f9f7b0e72 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be17fe607df7473fa645208aaa6c4fb7 : SUCCESS in 2m 50s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/57aa0a56fdde475cb406ae2495ffd5b7 : SUCCESS in 42m 48s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/56a2cc777bf549a4b5b492ad8e0e85b7 : SUCCESS in 1h 03m 25s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/5f3635da0d1b4533a0a580816f816d47 : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"48cc591dec28def80b3445282cda68d1d550ef64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 16:42:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cde9888a13e9e18fd9f2695da9d8e3878004e1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 17:38:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/417cb90aa51646eb8168542c9a61a8d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85f07129abd4f59aadc57933d922c03 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c41088458345a39d3e0739c8c334f7 : SUCCESS in 3m 00s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/18833d3d697c4cbabadef0867986325a : SUCCESS in 38m 44s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/25cce46d999a4ae29d5d081e1e15c2ba : SUCCESS in 48m 28s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/48b7a956c04f4556a9d34dee809cc678 : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9f9ad3a5ab6692bc365c7e1e23fa5520ab9374b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 17:38:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2e71b0b74dd2665227a5ca030b8449b3c5d95d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 17:39:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87241593ece9478c8a8ffc286128b263\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c66006bf8cb454498e76b1199da412a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e335e6ddba001537776412b362ff0629d4a56ff","revisions":{"0e335e6ddba001537776412b362ff0629d4a56ff":{"kind":"REWORK","_number":1,"created":"2023-06-01 15:25:54.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/64/885064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/64/885064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/885064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/885064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/885064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/64/885064/1"}}},"commit":{"parents":[{"commit":"47c5afe79aa334b300b9f15f3f23ff76d812d742","subject":"Remove sphinx doc building integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/47c5afe79aa334b300b9f15f3f23ff76d812d742"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-06-01 15:24:37.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-06-01 15:24:37.000000000","tz":-420},"subject":"Add release note for build_sphinx removal","message":"Add release note for build_sphinx removal\n\nThis wasn\u0027t done in the change that removed the feature but seems like a\ngood idea.\n\nChange-Id: I97bdda4ce89b26c6e99de6f1d6ad7d734cac9041\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0e335e6ddba001537776412b362ff0629d4a56ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0e335e6ddba001537776412b362ff0629d4a56ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
