)]}'
{"id":"openstack%2Fpbr~899848","triplet_id":"openstack%2Fpbr~master~I32f974db37e364cf634b050b40bf0820dce0a3a3","project":"openstack/pbr","branch":"master","topic":"pep-660","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-11-01 23:28:22.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-01 23:28:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I32f974db37e364cf634b050b40bf0820dce0a3a3","subject":"Add support for PEP-660","status":"MERGED","created":"2023-11-01 11:18:21.000000000","updated":"2023-11-01 23:29:17.000000000","submitted":"2023-11-01 23:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899848-pep-660","meta_rev_id":"0a598e1f49cb51c3657901e4bdeef37779699960","_number":899848,"virtual_id_number":899848,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-01 23:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-11-01 19:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-01 22:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-01 22:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-01 12:58:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-01 13:37:00.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-11-01 15:10:55.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-11-01 22:06:59.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"b5921adcc2faaae2cd79527b4fb5a762c69c0d55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-01 11:18:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"283b76fd3179346c1463f3eb4fc36437647a9af0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 12:58:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51fa43812836497dae80ae1bb27bcd34\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ecf0c19e216e40eaa939e6f0452e35ba : FAILURE in 24m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2a0ac4002e34160a949e7b9e008db1d : SUCCESS in 1h 32m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902376d0f22f4b359bd74be3ec07efb7 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81174c8cc79b4348af4590342c5dd87b : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16ee1640cc114be6bd900863fbd64861 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fdc01139a3f4739bf0c6781bdc75308 : FAILURE in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a61a0318a6245299e5b0880e8f1de93 : FAILURE in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a7f7c2ff10843249a2ca44da4377282 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a329a9427e0e41b2b36644e4807496e5 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a94b0ccacbaf47ebbaafbdac834499da : SUCCESS in 4m 04s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/fd18f98b084647818b068f5ab1a9d40d : FAILURE in 29m 28s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/ece3f676c4634c41a7648a7add6fafca : FAILURE in 41m 33s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/1e05cdc0b4274fbdaa49114b16df93b4 : FAILURE in 37m 10s","accounts_in_message":[],"_revision_number":1},{"id":"afc309515149a064cbf14c9e43956f84fe577573","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-01 13:37:00.000000000","message":"Patch Set 1:\n\nAt a minimum, we\u0027re going to have to do some sort of dance to exclude PEP 660 support on Python 2.7 and 3.6 due to needing a newer SetupTools than supports them (though the 2.7 tests are actually choking on a syntax error when trying to use importlib_resources). We could still break PEP 517 testing out separate for those, I suppose, though I don\u0027t know that it\u0027s worth it.\n\nIt\u0027s not immediately clear to me why pbr.tests.test_integration.TestIntegration.test_integration(nova) is failing in the install jobs. Seems like it\u0027s not finding migrate.cfg installed in the venv but I don\u0027t see obvious errors indicating the reason. Pip at least thinks it successfully installed nova but we might need to coax some additional debugging output from it and/or capture a copy of the built nova-0.0-py3-none-any.whl file.\n\nThe tempest test is erroring trying to install master branch neutron with Python 2.7 and pip 9. I\u0027m not familiar enough with the job to be sure this is intentional.","accounts_in_message":[],"_revision_number":1},{"id":"a3d0a69d716e2656f4b2581c7f66c290a3b8d615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-01 14:01:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"788f180050ee0ff6218dda6627f35dcbe180bc7b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-11-01 15:10:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dbbfceb9bad2376152da8edb79221b8593a6313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 15:49:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/859b1f3c21434516916e3f7c03e9a05c\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/fe471a5ef7d74e338e6ce4e8748bf891 : FAILURE in 24m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c89360bc5ed44796b0582710c43eabaf : SUCCESS in 1h 39m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a557d57bb74dd98c161a7d427eb657 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8381af2e012e42eabdd205c9087ad29b : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7781f4ebe76c486a8cfdd346ee8b65cf : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d364d26f12f478090be39cebf5f5fcc : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22af7a6ad2564fe79649e12a0fcc012a : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a4a2b19e3094348ac4b3be381047e2f : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc8ecf12ee794c4589edfb84dc8ae3f5 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3498dd2ec6464589b8d402c433983997 : SUCCESS in 3m 50s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/612063cbee7e4e7cbd506079c7cde50b : SUCCESS in 34m 19s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/9670da7143264186ab3aa8227f871cf9 : SUCCESS in 38m 41s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/74c958a03be04986a7f1aabbbdf3d6a6 : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e0da8373a098ef0e27cdb8de85359d7e141ec651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-11-01 19:53:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c1e6fad7442935e9a76702ecc639d27c0d394820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 21:49:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81982b1cd21e44858b3c11c891bf0c05\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3bf26ebe65e44fa3bfee1068339c6f8a : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66ba28e808604de780176f779cdc6155 : SUCCESS in 1h 48m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83df5261f47a4fbfa89d02ec05fbd1f5 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb19907281f643eba0aee0ca009ab645 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eecc65a033da4f0a8b270c2079dedb59 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8235bbc3a5164c2c8be8766222f19115 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4f78f37eea04f3189d3349059710428 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dce9ad0ab394b4aa885756a244f8856 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/644e226dbcbe4bb098a08c40536e2093 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/513b081f2c4b42c8a93dd51d8dae82fa : SUCCESS in 5m 42s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/38fe1559a8ca4ab8b9dad93e732ed065 : SUCCESS in 28m 39s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/0f60b587db9c41dabbae9f2305c3a0d7 : SUCCESS in 38m 22s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/ef36469614a1402593e22f3bf11fa772 : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":3},{"id":"31c943699b2d7b96334d77ec77e7c1b920c741a1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-01 22:06:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"465c1f070516286b09047d8c057d57790b32016f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 22:08:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"82ed0aeaf8fafb4cb02b9d201041bb0c66c4205d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 23:28:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600ef2ab2a1842d0922274019a9f828a\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/81a760ad7e454f39933c23c577649227 : SUCCESS in 43m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88136e7ff9044a4592fb17ce317da8bc : SUCCESS in 1h 17m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60fc3f0d1d804bbb934fbf80fbe2f59b : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108a8027f141450a9798b640b3e53029 : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b4054e4b2344a25accb2b0e8fa90862 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25233e6fc9c14afe9e95bc4d3fe88365 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e973dc141e2d4b358914eda18944c0c2 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f91fd4ba41a41d3ab1baea6e639bade : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2422fd1453654554a6d842a5d0153914 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c53a6fa2b514620a3dd09eb46429729 : SUCCESS in 3m 25s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/72616ce4a0d446d5a8321665e2d931d0 : SUCCESS in 23m 50s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/524453d029d74aa2aa4afcb7b9028e04 : SUCCESS in 26m 40s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/7f780f14ee7d447295758b882365f5bc : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":3},{"id":"b4b42f80ce420c9d2bb259b959366b799a689af9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 23:28:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0a598e1f49cb51c3657901e4bdeef37779699960","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 23:29:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c1e3e481834ba4a97f8543b5d2d449\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e36ebf13264fdcbfb5bbdc4ac711b9 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"457a89ffda32cb7a49fa680a71d8c043bd23b352","revisions":{"3b89be7a4362cafb19981aeaef777ba55a941d42":{"kind":"REWORK","_number":1,"created":"2023-11-01 11:18:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/899848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/48/899848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/48/899848/1"}}},"commit":{"parents":[{"commit":"d03d617c09e7ba8ddf62d1e53d71685cd708e2da","subject":"Replace imp with importlib.machinery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d03d617c09e7ba8ddf62d1e53d71685cd708e2da"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-11-01 11:12:50.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-11-01 11:18:09.000000000","tz":0},"subject":"Add support for PEP-660","message":"Add support for PEP-660\n\nAs with PEP-517 support, this is simply a shim over the setuptools\nimplementation added in setuptools v64.0.0 [1].\n\n[1] https://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I32f974db37e364cf634b050b40bf0820dce0a3a3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3b89be7a4362cafb19981aeaef777ba55a941d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3b89be7a4362cafb19981aeaef777ba55a941d42"}]},"branch":"refs/heads/master"},"bb4bbf5f4ac8850658c0e608996193db7ed4be29":{"kind":"REWORK","_number":2,"created":"2023-11-01 14:01:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/899848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/48/899848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/48/899848/2"}}},"commit":{"parents":[{"commit":"0f77bf8b9be20376d52c12e45895d7b0adb35d53","subject":"Update integration test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0f77bf8b9be20376d52c12e45895d7b0adb35d53"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-11-01 11:12:50.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-11-01 14:01:21.000000000","tz":0},"subject":"Add support for PEP-660","message":"Add support for PEP-660\n\nAs with PEP-517 support, this is simply a shim over the setuptools\nimplementation added in setuptools v64.0.0 [1].\n\n[1] https://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I32f974db37e364cf634b050b40bf0820dce0a3a3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bb4bbf5f4ac8850658c0e608996193db7ed4be29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bb4bbf5f4ac8850658c0e608996193db7ed4be29"}]},"branch":"refs/heads/master"},"457a89ffda32cb7a49fa680a71d8c043bd23b352":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-01 19:53:30.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/48/899848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/48/899848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/48/899848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/48/899848/3"}}},"commit":{"parents":[{"commit":"6df16432aad01f0bd8b219a5e8a0a12062f10985","subject":"Fix PBR integration testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6df16432aad01f0bd8b219a5e8a0a12062f10985"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-11-01 11:12:50.000000000","tz":0},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-11-01 19:53:18.000000000","tz":-420},"subject":"Add support for PEP-660","message":"Add support for PEP-660\n\nAs with PEP-517 support, this is simply a shim over the setuptools\nimplementation added in setuptools v64.0.0 [1].\n\n[1] https://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I32f974db37e364cf634b050b40bf0820dce0a3a3\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/457a89ffda32cb7a49fa680a71d8c043bd23b352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/457a89ffda32cb7a49fa680a71d8c043bd23b352"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
