)]}'
{"id":"openstack%2Fpbr~900728","triplet_id":"openstack%2Fpbr~master~I6a66b6f687ec24b3ed7208613c0970dbb460a43b","project":"openstack/pbr","branch":"master","topic":"editable-install","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-12-19 17:34:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6a66b6f687ec24b3ed7208613c0970dbb460a43b","subject":"demonstrate wsgi pyproject.toml bug","status":"ABANDONED","created":"2023-11-11 21:47:37.000000000","updated":"2023-12-19 17:34:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"669d58a462f93692767864c2a28c3b13d7e57dca","_number":900728,"virtual_id_number":900728,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-11 22:56:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-11 22:56:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29cf79db41ea134303d22d607e7133da72685e18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-11-11 21:47:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0354c816cc88af68f8d1ba0f96ebe4910b76363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 22:56:00.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65984215b962456cbce66c27cefcda56\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/a3d720a3b4a647d2927d48e65db7e609 : SUCCESS in 59m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18bb4fe69e9140699b2ef1aa5ec4ff43 : SUCCESS in 58m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af1f514bc524d66bd425099584cc03e : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb4ee2534a14fb9bf064e519be376ea : FAILURE in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/092c2987fa4b417d96f748f18739755a : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ada2db92427412ca3af629455660296 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9078ce2ff1684010867f20d35602b2db : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c976927b28c4362b22b25e01200a962 : FAILURE in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc0394f9499d490a8b7f487afefab9ae : FAILURE in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f252733f0a3e432ea4efd4a7b588c9e7 : SUCCESS in 3m 59s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/92d23b1fd30a4fae904b9300b5cf51fa : SUCCESS in 25m 33s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/8e7c564d74144a7faa907ccb5472791f : FAILURE in 25m 42s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/ff5baf35a5bd4a85b632291c649327c4 : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"669d58a462f93692767864c2a28c3b13d7e57dca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-12-19 17:34:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25b116c214eda596c1c291df54810b69489f82d0","revisions":{"25b116c214eda596c1c291df54810b69489f82d0":{"kind":"REWORK","_number":1,"created":"2023-11-11 21:47:37.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/28/900728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/28/900728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/900728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/900728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/900728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/28/900728/1"}}},"commit":{"parents":[{"commit":"457a89ffda32cb7a49fa680a71d8c043bd23b352","subject":"Add support for PEP-660","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/457a89ffda32cb7a49fa680a71d8c043bd23b352"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2023-11-11 20:19:14.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2023-11-11 21:47:32.000000000","tz":0},"subject":"demonstrate wsgi pyproject.toml bug","message":"demonstrate wsgi pyproject.toml bug\n\nThis change asserts that when a wheel is built with pip\nwheel wsgi scripts are generated with or without the\npyproject.toml\n\nThis change also shows that they are not generated when\nthe whell is build via pip install -e.\n\nChange-Id: I6a66b6f687ec24b3ed7208613c0970dbb460a43b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/25b116c214eda596c1c291df54810b69489f82d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/25b116c214eda596c1c291df54810b69489f82d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
