)]}'
{"id":"openstack%2Fpbr~909581","triplet_id":"openstack%2Fpbr~master~I63daac9345cd60daf01e7c6dd67d4182222eb699","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-03-19 08:01:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I63daac9345cd60daf01e7c6dd67d4182222eb699","subject":"Remove six","status":"ABANDONED","created":"2024-02-20 14:01:05.000000000","updated":"2024-03-19 08:01:08.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94cad466a8bb402da3756ee7bfda78a5d29dfedb","_number":909581,"virtual_id_number":909581,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-20 15:23:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":-2,"date":"2024-03-04 15:44:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-20 15:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-04 15:44:10.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"21bfb9816a5462ed7356abed7f52141e8a86ff93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-02-20 14:01:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20c3f6eaea6b60f00277bf7139f2ed4b9ab67ed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 15:23:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b295ed3eea34acf9b6be87849db51fd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/005cd07f19494d439200b900bab66efa : SUCCESS in 1m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e6ace67593c849238aa3ffac3adc4c38 : FAILURE in 9m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/851cbe771b364c35914e6f1a572b59e2 : SUCCESS in 1h 19m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268f307dfa5e4ce7aac243fb907203fe : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c1e2e43dcc40e99e23e7bb41ddec90 : SUCCESS in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d7898c0e784bffb31db2d48edead57 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bd5233ce83e43428b5bda48a4687fcc : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c2c5181247143809f4b8ce6a4a12a83 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c3585744a794b5b99abcaf50c433ddc : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0264b4773fa34e969f99cf70d02a7c18 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66c73ee8f8b8495d8e7b05c9f715adb9 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3a399aacac84272a6a1a0c9ce27fc76 : SUCCESS in 4m 21s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/b6443e42bed5483c96c164248ffe8141 : SUCCESS in 36m 31s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/48fbabaee83e4040a7496f77ca80e40b : SUCCESS in 32m 19s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/25ebec7b51e84feba48699f9f86e9cc9 : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1d0e4c34cca007178ce975547caa27480f7c3634","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-03-04 15:44:10.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94cad466a8bb402da3756ee7bfda78a5d29dfedb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-03-19 08:01:08.000000000","message":"Abandoned\n\nWe still need python2 support.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9eb188838ea13f486bdee09f7eddc384f608abf","revisions":{"c9eb188838ea13f486bdee09f7eddc384f608abf":{"kind":"REWORK","_number":1,"created":"2024-02-20 14:01:05.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/81/909581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/81/909581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/81/909581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/81/909581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/81/909581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/81/909581/1"}}},"commit":{"parents":[{"commit":"f55524c08e87eabc6877016e7e75a2d55f0bb098","subject":"Merge \"Use summary for the package summary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f55524c08e87eabc6877016e7e75a2d55f0bb098"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-20 13:59:15.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-02-20 14:01:03.000000000","tz":60},"subject":"Remove six","message":"Remove six\n\nThis patch removes the use of six, and therefore, removing\nPython 2 compat. Only Swift is still using py2, but there\u0027s\nalready some attemps at removing it.\n\nChange-Id: I63daac9345cd60daf01e7c6dd67d4182222eb699\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c9eb188838ea13f486bdee09f7eddc384f608abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c9eb188838ea13f486bdee09f7eddc384f608abf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
