)]}'
{"id":"openstack%2Fpbr~919311","triplet_id":"openstack%2Fpbr~master~Iec826ae9a089509baa61e6c8b09ca4d152ef4f83","project":"openstack/pbr","branch":"master","topic":"bump_sphinx","hashtags":[],"change_id":"Iec826ae9a089509baa61e6c8b09ca4d152ef4f83","subject":"[DNM] Testing docs bump with new Sphinx","status":"ABANDONED","created":"2024-05-10 20:57:46.000000000","updated":"2024-05-14 15:19:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"8cf3ae87abefdac62bf1e12d6f04b52c13430d41","_number":919311,"virtual_id_number":919311,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-10 21:42:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-05-10 21:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc1e3ee879b84e9f3f6a1d800524071697d58396","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-10 20:57:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10a8153e3788ed6db373698c0eb5f00ea3417e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 21:42:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5e0f071db8b4e238da24aa17d6ef6f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3fc5db73614b9196332fc0707a89e2 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf13abee7f84999a54e37b20866cc2e : SUCCESS in 2m 53s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/ae2365211d394782b736a24247cba930 : SUCCESS in 32m 49s\n- pbr-installation-openstack-focal https://zuul.opendev.org/t/openstack/build/b37dccfe492e40a59f76b65178db5998 : FAILURE in 26m 38s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/dc17338a09d148ee91d5ad11d2966ff9 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8cf3ae87abefdac62bf1e12d6f04b52c13430d41","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-05-14 15:19:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edf135c11fb29caf6fd5c3b7dbe876c7505e14f1","revisions":{"edf135c11fb29caf6fd5c3b7dbe876c7505e14f1":{"kind":"REWORK","_number":1,"created":"2024-05-10 20:57:46.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/11/919311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/11/919311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/919311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/919311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/919311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/11/919311/1"}}},"commit":{"parents":[{"commit":"f75e1056f1763b567d7044ec63279afe110aa75e","subject":"Omnibus PBR CI fixups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f75e1056f1763b567d7044ec63279afe110aa75e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-10 20:57:44.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2024-05-10 20:57:44.000000000","tz":-300},"subject":"[DNM] Testing docs bump with new Sphinx","message":"[DNM] Testing docs bump with new Sphinx\n\nSphinx updates can be scary.  This change leverages a constraints bump to see what breaks.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/916565\nChange-Id: Iec826ae9a089509baa61e6c8b09ca4d152ef4f83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/edf135c11fb29caf6fd5c3b7dbe876c7505e14f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/edf135c11fb29caf6fd5c3b7dbe876c7505e14f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
