)]}'
{"id":"openstack%2Fpbr~919533","triplet_id":"openstack%2Fpbr~master~I7c80e93627597101cbed64a0efdc18140dbc854a","project":"openstack/pbr","branch":"master","topic":"drop-focal-integration-testing","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-05-14 11:51:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c80e93627597101cbed64a0efdc18140dbc854a","subject":"Drop Focal integration testing","status":"MERGED","created":"2024-05-13 15:30:40.000000000","updated":"2024-05-14 11:53:54.000000000","submitted":"2024-05-14 11:51:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919533-drop-focal-integration-testing","meta_rev_id":"98c5018045a37b8e8a649f92ccf6526347f587d4","_number":919533,"virtual_id_number":919533,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 11:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2024-05-14 10:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-13 16:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2024-05-14 10:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-13 16:39:42.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-05-13 17:18:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-14 10:08:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a7ccf5d20fc1b4e0a5d01e61db56ae22d22070ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-13 15:30:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f23edb235c5914a608210592671ee7db282d1f9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-05-13 16:39:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bcffd6df250a76079b919ac52578ef9d687ce220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:18:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16f8242f23f47308abc7880a24a18b8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da7f8c00d3fb47219aa9a890f2ac9e2f : SUCCESS in 1h 39m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1b9914227f4df2828ddcbd2b695084 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a6840864134e38b0b423de96d1405d : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adfcb65609d44f108a35a19d044f3957 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a92b0693412b4b598b05e6747a6d6060 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d99bea2ff564652a84d1c3e61618591 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fde302dffefc4ca7ad3426eb154c5716 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53e2dc1e90d242db8682e31fe434634c : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24a5cee8e21c46d099fbbc5423032294 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59d92d7c0ebf4a06a07de930ce1875b9 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/caef7caa50824ad8b4cf2377ed701882 : SUCCESS in 4m 35s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/4d80e075edb4483eabadc25aacb888a7 : SUCCESS in 37m 33s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/11115f4c379a4224b70cb709eb24428a : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b680520045c1be85d27d56ea15b3ea77e17309d5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-05-14 10:08:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f67f0a0c6a764b52b121d98b4dd41b9d2a731e64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 10:10:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f48675eecaa47386f0ec366517d951c458b5115e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 11:51:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6af4a7f400f4179901c400b7aa31fe7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e5a7034076a4d48b5459ce969e7a141 : SUCCESS in 1h 34m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa36d7de1ba4d92bd69a8804c5e1077 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7224f338a774adf8c4e556f0dc9d899 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/977ea70a60054635b33ded5287bfac85 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/864d577bfbdb4e448ce37cdeff1fe14d : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b0142f16ad74163a362a10e4e66d41b : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8ae1dbd95574a87801b7f8ffa23f8c5 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/917358bda83c4175ba5096125338a92a : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b4a4eecd9484a00a0acaa841abd6a23 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6581a69077994570968692e45d1aef4c : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba221bb88aeb4ebb9f2ecf334f124def : SUCCESS in 5m 43s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/cc8cc9af81684a98baef9f3088e4a47b : SUCCESS in 30m 49s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/49bc3691dde34887aae7156f808dec20 : SUCCESS in 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6fe59f21f13fb766001e653abe80c4d0abc86bb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 11:51:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98c5018045a37b8e8a649f92ccf6526347f587d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 11:53:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebac40ed4fb9431dbca1377fa9a57334\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f45411806544a892d3bbeb203f86d8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb56c72f832d4be2907207757ac7eb171715e151","revisions":{"bb56c72f832d4be2907207757ac7eb171715e151":{"kind":"REWORK","_number":1,"created":"2024-05-13 15:30:40.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/33/919533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/33/919533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/919533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/919533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/33/919533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/33/919533/1"}}},"commit":{"parents":[{"commit":"f75e1056f1763b567d7044ec63279afe110aa75e","subject":"Omnibus PBR CI fixups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f75e1056f1763b567d7044ec63279afe110aa75e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-13 15:27:04.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-13 15:30:34.000000000","tz":-420},"subject":"Drop Focal integration testing","message":"Drop Focal integration testing\n\nThere are projects which no longer support python3.8 in our integration\ntest. This means they cannot be installed on Ubuntu Focal any longer\nwithout extra steps like checking out old eol tags. Rather than try and\nkeep up with this list (that will only grow over time) we drop the Focal\nintegration test and simply rely on unittesting for python3.8 and older.\n\nThis is in line with previous decisions made for python2.7 testing. It\nisn\u0027t ideal, but should still provide sufficient coverage for the most\nglaring incompatibilities we may try to introduce.\n\nChange-Id: I7c80e93627597101cbed64a0efdc18140dbc854a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bb56c72f832d4be2907207757ac7eb171715e151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bb56c72f832d4be2907207757ac7eb171715e151"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
