)]}'
{"id":"openstack%2Fpbr~920594","triplet_id":"openstack%2Fpbr~master~I240e6c7bbe471ec3a6f6a2a4c93f488709414ea4","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-05-28 00:47:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I240e6c7bbe471ec3a6f6a2a4c93f488709414ea4","subject":"zuul: Drop retired repos from required-projects","status":"MERGED","created":"2024-05-27 20:43:58.000000000","updated":"2024-05-28 00:49:18.000000000","submitted":"2024-05-28 00:47:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920594","meta_rev_id":"94435e61098b27d5eb60f32c563b4c6033fe0538","_number":920594,"virtual_id_number":920594,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 00:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-27 22:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-27 22:17:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-27 22:17:27.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2024-05-27 22:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6584c4d983d23f3c2ec19ea9ad4f9883fe4e6d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-27 20:43:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d571a53c4d692d8b6a55d09e359ff7f4b364f5a0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-05-27 22:17:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46530442bbc50eb358dc0455ee3f76d9330d0331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 22:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3a51376099e4f53a32546143eab324b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/093e06099b5a4a1280f187a6e8ff1330 : SUCCESS in 1h 55m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffac421fc6b4424e8bd8ef64a0c5f195 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ad850006bf4e9da612b029bc511cbf : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85f59eebbda846b0a0f8df72a778bd3b : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19397a12892b4e42b590ee6ed55ed0d7 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33da1e681a124cc688f01d14f83a0141 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e6585ef4ba84e49b2687b7704e1068d : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1702344833b84ab4a5914651b9088971 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2fa4844db7247cbabac46171de4ff83 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5624cc926704c699d3c8df00fe6a6c1 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49526542faa143d1bd84ed216a5a9b7b : SUCCESS in 4m 41s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/9262e0bfb8554fe9a2cd2f7cd548a571 : SUCCESS in 29m 10s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/36d1575162e143c1bc9f6134ae6d8a4b : SUCCESS in 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"91a639540783d8d6b632885752be74ea0899417e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 22:48:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec9dcf9edd821593725d2f3663c78f95bdf974c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 00:47:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b3c6acead93411ca8c17547c604113b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ca8adecfffc4287a2d17ebe69e8d74a : SUCCESS in 1h 49m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051da5885087499eb6c0d29f5bf51c56 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad5fcc5d9aa441b8eef4a31b5ef899e : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9389d619421d41708479685313152410 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0da18395ea6491cb46c0a72e9d8fb7e : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f027e87d16294a1db9fb30df5ce6d92d : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddc539bb03b24fa8862bc8392c4f3511 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0599e8f99a134d208395fb8be63bd038 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/330275517859445983cc65dc2dfeb2d8 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11c9e05551fb4ff9b66ec7f2823e0043 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13dfe76251984143944c91f93d9d13ef : SUCCESS in 3m 53s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/efb518bdf283427e840d288cbd051554 : SUCCESS in 26m 03s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/fd7fb741f8d64dc399ef53a748c0e785 : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e83dd9441156804108a3bc7a35a13e067d537311","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 00:47:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94435e61098b27d5eb60f32c563b4c6033fe0538","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 00:49:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37024591b2344c698cce3ed702ef71fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb4d0c015f343eaa31b5c3c62a92f3a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"240b176e01f8670cdf7a5eeac774927181b45c9e","revisions":{"240b176e01f8670cdf7a5eeac774927181b45c9e":{"kind":"REWORK","_number":1,"created":"2024-05-27 20:43:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/94/920594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/94/920594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/94/920594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/94/920594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/94/920594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/94/920594/1"}}},"commit":{"parents":[{"commit":"bb56c72f832d4be2907207757ac7eb171715e151","subject":"Drop Focal integration testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bb56c72f832d4be2907207757ac7eb171715e151"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-27 20:40:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-27 20:40:55.000000000","tz":120},"subject":"zuul: Drop retired repos from required-projects","message":"zuul: Drop retired repos from required-projects\n\nThese repos have been retired and are no longer available:\n\n- openstack/python-saharaclient\n- openstack/sahara\n- openstack/sahara-dashboard\n\nChange-Id: I240e6c7bbe471ec3a6f6a2a4c93f488709414ea4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/240b176e01f8670cdf7a5eeac774927181b45c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/240b176e01f8670cdf7a5eeac774927181b45c9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
