)]}'
{"id":"openstack%2Fpbr~920676","triplet_id":"openstack%2Fpbr~master~I41254afddfdb68780dc8989b64ec0527587c19be","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-05-28 20:01:21.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-05-28 20:01:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41254afddfdb68780dc8989b64ec0527587c19be","subject":"Use SetupTools\u0027 vendored distutils in tests","status":"MERGED","created":"2024-05-28 14:58:15.000000000","updated":"2024-05-28 20:02:55.000000000","submitted":"2024-05-28 20:01:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920676","meta_rev_id":"0d55444a738dfcdbcc2f798f74965834ea9baeb3","_number":920676,"virtual_id_number":920676,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 20:01:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-05-28 15:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2024-05-28 18:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2024-05-28 18:02:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-28 14:59:30.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-05-28 17:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-28 18:02:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"194da950f779236c33657c6a8dd3dd648c389bdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-05-28 14:58:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75de0d26526c1152364d2a41c501fc8848a543b8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-28 14:59:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37d50d94b0edac0dc98497d35db7a9f023d27dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-05-28 15:59:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"07b01e3ed1f9712452808b03e4666389c0b1c0bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:41:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29c6a75713f4562b0f531215ae7d904\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d9d2b1337b246b6a7b3e8d51f6f59b3 : SUCCESS in 1h 35m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa9cf599621451d933ec408bdcd3ebc : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be5384c6382e41faabdc8b47e51525d9 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46eb16d26e5543848634854d295b66df : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c871e7e9bd84593a39dba48413fd4a7 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5017351fc4124b08bfd93c8de2e06ad4 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f6bd1c6532049868bd0c14de6ed7426 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d0df78d14a945fab3912ab0b9263328 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0290a1ce1044d44bda4bf6a73f1ad7c : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c410000aa2148dbacbc27302ac0d6a9 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c51a6c3026647869cbc4749098b9627 : SUCCESS in 5m 23s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/878a8ee642cc4e839e5848656fa5eae3 : SUCCESS in 28m 29s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/1492df3800d240bbbf1a832fe6d5178e : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a230608e5bfd715d781d81d9196d1689096c212d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-28 18:02:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6b0a4022c5181752cf1f200f14c2bc840b97f293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:04:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f92e17ccfb52f0f59bdc7071629db5932e851183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 20:01:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72722386f64844bcb47def1b43eee45e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c15359c7c39447d4baf834ec6ca7a638 : SUCCESS in 1h 40m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3d5b584847406b86886a986e82bb79 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c0751b72e1408994070a92661e8bfd : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c808ae5f1102430e975d1bd5f56d053f : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbe42245f3604b769ade6509937f6d6b : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6f2cd8f70b24a17bf366a974be5e522 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0fee59808be4bbf86adabccf26a4150 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e00bc80dca417cb4eb923a2642f689 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/077f9314bb244a6b94f34fdea2f2f083 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/995dffa3583b45809ff448f2f9e31397 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41403b32012149979a44aa8a26a439c3 : SUCCESS in 4m 42s\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/6528610f6656447f9119a772150b4cc2 : SUCCESS in 31m 54s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/62b96d19dec4427d99cadbc5fdde1991 : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2e670cec96fb4b8b2cc1ab247df0196f088b6fd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 20:01:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0d55444a738dfcdbcc2f798f74965834ea9baeb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 20:02:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c681d4e7ce1d483fae9a5996f029d772\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859ee0773e284a54a64ed2b354c5c6fa : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"203f33cdf4f7d7c6c225898052577f1c25a38593","revisions":{"4c43764d7801f97bbe8cc57a9ee30aa97055e5ef":{"kind":"REWORK","_number":1,"created":"2024-05-28 14:58:15.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/76/920676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/76/920676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/76/920676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/76/920676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/76/920676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/76/920676/1"}}},"commit":{"parents":[{"commit":"240b176e01f8670cdf7a5eeac774927181b45c9e","subject":"zuul: Drop retired repos from required-projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/240b176e01f8670cdf7a5eeac774927181b45c9e"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-05-28 14:54:17.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-05-28 14:58:04.000000000","tz":0},"subject":"Use SetupTools\u0027 vendored distutils in tests","message":"Use SetupTools\u0027 vendored distutils in tests\n\nDrop a temporary workaround we put in place for a SetupTools bug\nthat was fixed over two years ago. This becomes necessary as we\nprepare for testing under Python 3.12, since its stdlib no longer\nprovides the distutils module, so we actually need the one from\nSetupTools for it.\n\nChange-Id: I41254afddfdb68780dc8989b64ec0527587c19be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4c43764d7801f97bbe8cc57a9ee30aa97055e5ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4c43764d7801f97bbe8cc57a9ee30aa97055e5ef"}]},"branch":"refs/heads/master"},"203f33cdf4f7d7c6c225898052577f1c25a38593":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-28 15:59:46.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/76/920676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/76/920676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/76/920676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/76/920676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/76/920676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/76/920676/2"}}},"commit":{"parents":[{"commit":"bcc1f18440336446b2d0ae67080e59e6523e410d","subject":"Also run coverage tests serially","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bcc1f18440336446b2d0ae67080e59e6523e410d"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2024-05-28 14:54:17.000000000","tz":0},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-05-28 15:59:40.000000000","tz":-420},"subject":"Use SetupTools\u0027 vendored distutils in tests","message":"Use SetupTools\u0027 vendored distutils in tests\n\nDrop a temporary workaround we put in place for a SetupTools bug\nthat was fixed over two years ago. This becomes necessary as we\nprepare for testing under Python 3.12, since its stdlib no longer\nprovides the distutils module, so we actually need the one from\nSetupTools for it.\n\nChange-Id: I41254afddfdb68780dc8989b64ec0527587c19be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/203f33cdf4f7d7c6c225898052577f1c25a38593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/203f33cdf4f7d7c6c225898052577f1c25a38593"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
