)]}'
{"id":"openstack%2Fpbr~924216","triplet_id":"openstack%2Fpbr~master~I3f4e9f95d10dbbba8499ac4354ff2bf402befd12","project":"openstack/pbr","branch":"master","topic":"add_py312_support","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-05-04 07:50:28.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-22 08:10:34.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2025-05-04 07:50:28.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-05-04 07:50:28.000000000","reason":"Change was abandoned"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-05-04 07:50:28.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-25 16:56:22.000000000","reason":"Reviewer/Cc was removed"}},"hashtags":[],"change_id":"I3f4e9f95d10dbbba8499ac4354ff2bf402befd12","subject":"add python 3.12 support","status":"ABANDONED","created":"2024-07-16 11:23:51.000000000","updated":"2025-05-04 07:50:28.000000000","total_comment_count":30,"unresolved_comment_count":10,"has_review_started":true,"meta_rev_id":"e580d9bf9c5ee467430052d590adaff48de9ccdc","_number":924216,"virtual_id_number":924216,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-22 09:56:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":-1,"date":"2024-12-17 18:55:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 12:57:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-18 16:45:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2024-07-18 18:53:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-09-25 16:56:22.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REMOVED"},{"updated":"2024-11-18 08:48:16.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2024-11-25 21:48:49.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2024-11-26 02:52:24.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2024-11-26 02:52:24.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-11-26 10:39:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-12-17 18:55:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"7f80acc9d939e518cea962b992a27ada095450e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-16 11:23:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2939d230770e542a9e9869f1c118add544874379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:57:52.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ed000a63d7448ca932f461c7147d17a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bc56f2af8e24a6aaa39369b5e543b00 : SUCCESS in 1h 28m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678e24f8e6f5474b850424a9880cfa0d : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508311bc54c64653a9d86f92f9382654 : FAILURE in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb02aa99dd440589a3d6db1b94cbbd0 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2cb19d5d5bea4f31aad794b01a270e7a : FAILURE in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb6aef3f3af43e0ac67ee94bcb08685 : FAILURE in 2m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5af60ccdc6704c249d4e9f7027c34afc : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ff584e1694b494abcf07d78b6a5bf54 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f8d6173eb3c4c8394cbdd2638c11b5a : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6818582479044068ae25861df757b91e : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1874077605b54d2083f7e79ba0df80b6 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9b30dbad03243d0b2a7a17abcb0ebd1 : SUCCESS in 7m 51s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/71f41ccda9104074a1c2c627c28c0d5a : SUCCESS in 26m 51s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/063ba049af264eaa9a1220760265f988 : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"687300618549edf699cc24b041cdc248abc78a63","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-18 16:45:27.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2a4f8919d999e05ce5909b1647201109ef4510e0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-18 18:53:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ffae1b4384d5bc3cc82f2e233b594888f76a2a9","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-29 08:42:20.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d963d0a3d02277efb6a86879ab75cd4ad810802e","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-25 16:56:22.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_15343\u003e.","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":1},{"id":"0abb43d869c2d9a006046d019e5b01bdfad6f4eb","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-18 08:48:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a468672a552992c67ab8ef0bdf507711b10376f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 10:41:35.000000000","message":"Patch Set 1:\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34dbd993fe594509a09d9f6e27a5acfa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cb599f6650b4ed5ac14e27f9c088f5f : SUCCESS in 1h 45m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc957ed609024ae890e247b0811d935f : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3adddffbea4370816b29936b9a6a86 : FAILURE in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b935822317094400856767e0111844a1 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/754645a59d594f798d60bda46f838405 : FAILURE in 2m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b6a9382d27749bd9e81ebdfadaedadf : FAILURE in 1m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fa3d42bc7394e84b3c8e9dcb61ff8e9 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39c992dca1f542a3b5a90c7984e7d8bb : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e5faa81bcba4859bd4137c416d75019 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4795f689bb5453580452f36f8655efb : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b16f835ac4a4f31a224225ced36a1d3 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22070695cbbc4ce1bd31540c02327ccf : SUCCESS in 4m 16s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/3c28a57213794936a00add6c7ab51b1a : FAILURE in 3m 06s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/6f0a049dad1643ea83c6eefd022efce3 : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7c9be23ea28ffbe7c53f947f7818996f30bdd25b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-18 17:00:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f1f3a854cfc742764f9a3c67cf8c81e71fd7b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 18:50:53.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10a84352f41b4ceda0036e9bdd6134bd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be020cbdb11f4f60ade3f63ee7a40f51 : SUCCESS in 1h 43m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb705abb1fd4f208153809795d31e54 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f553d58d3c48d9b739434a1598ca91 : FAILURE in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ed3ffc700d048799b97e2fe4bc3e682 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9c9daa836ba4972aff79385e98b6c84 : FAILURE in 1m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b286e918ca47868d53d7e06eeb18b1 : FAILURE in 2m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d31571fc1b5147d68ee33b06688bfbea : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9aee891e81947f0a7ea492c4eee5343 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5deebba6550e4695a2d0da1e437d4534 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73c9222541734a9986e29cb59768c782 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10855038f7d84a3cb23666f2958b0a31 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6718fc741d8406aa61dd5e486df7806 : SUCCESS in 5m 06s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/6ce9a92e20d443c5817f8420fcb34207 : FAILURE in 3m 44s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/cd707df04aee482c9e92abab4a21706d : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4f1e19dd02fb5e0d73687660442eb0d747c8f1e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-18 20:40:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e904cfea89675b96816063e4b277ff97e562f1b5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-18 21:03:06.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"487410142350e1808488c321920b91e7cdc42861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 22:03:43.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e96c3b19deb643c5be4a03991638d0ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bc55ea9e74e4818b295efeff5eb9211 : SUCCESS in 1h 21m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f46bb73a4c674118a55fd378fd620d8f : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d4f8187cc243739b86fafef6fc4c62 : FAILURE in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34142f31b259477ea581b79dd7b3150d : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d4404fdf8304257a5ba5b16e665618f : FAILURE in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1635319a64a04d33a44cf1d159789696 : FAILURE in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/516b8e2e676a44fe96fad5d6fa07d775 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/603f2f10e7f14f28889c457fd144505e : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a18c8f7babb8496a93b00f835ab5b826 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fad8ffd05544f598e2dc7672cd26ad9 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6efc2a1cfe34aa08bf4d40feb013dcc : SUCCESS in 6m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/394c208051c5446b9a7a1e636dd0da9e : SUCCESS in 6m 05s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/0806f88c08d940288b1a5e0b49b95561 : FAILURE in 3m 10s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/2851e60bc09b4716bb6352d056e3bd6d : FAILURE in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"ca02518707af3f93acbb465ff550f1d5e9b17b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-19 09:38:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6a07ffb074d5cadf8bb14ba166c836e53d1042f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 11:24:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfb55dd1deb849799dd7657fcb548dc2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a61b9c7e7ef540ab9a9d203f2420cee3 : SUCCESS in 1h 38m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e46e33717a434763b3e9e4bca66139a9 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32aec35f0aa4b37b8bb7ab0bab3d69e : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a36d68d94cbe4feba630c258cc919a5a : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d054b4ffd6ed41c49ed9b74f1855460c : FAILURE in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e60cd00259749e0b93ce374479eb39d : FAILURE in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bda38c85a07b481081fbf34dfc92071c : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35a22ac2c0b24642aa873de3c89ca58b : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dfa188d305146e586f60749ce062183 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56e37df5385f41c5802f47972c2b040a : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0cfe70efde748bd9c05419eff63210d : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0b767b98265499a962ec811b1590529 : SUCCESS in 4m 17s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/26fef5ad49df4442bd2a4ebd3d679ee2 : FAILURE in 31m 23s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/858b597f74bb4917ac86bd31000ee145 : FAILURE in 20m 35s","accounts_in_message":[],"_revision_number":4},{"id":"c9b90728dd93247b9245bbafd4e819cde16c22a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-20 13:05:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a2097d64e2034a792ce5b8b736a23bf29b73a621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:26:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4259c1d6ddfd4f068955993c7bf17b41\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/930d4b78f6aa40918fa6a8f7a700e7fc : SUCCESS in 1h 18m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9a575d77d84ae99a551ad80876b998 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74fb673257c84602adf49a8c1925adb8 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/043d38d80c8f4e3e8b9bced959a13d28 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba74317972df4e5facbf401e445ef9e5 : FAILURE in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee8a07d5e3f547c39e93558c01e1e9d8 : FAILURE in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ba952f565ac439bb8f39eb0f28ce5e4 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b6078bc9aa44ae7bf488568a49915e2 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbb2618d1ebf4806b599bdca8b790304 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/caf8df606fc84e98992afeb78f2c9f7e : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7b162e54f8d461588d59654bf0221a3 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ff507f67794427eb98fc9daba1f1798 : SUCCESS in 3m 01s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/6ec15bc3eb0a40fc83779ad3c55e54e8 : FAILURE in 20m 14s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/74502f06d4e7414fa3aa81bec0578ea1 : FAILURE in 20m 58s","accounts_in_message":[],"_revision_number":5},{"id":"b39f3bc44802cd2d90c08812b4183326885c4489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-20 14:36:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cd297d541feba0b6f2b01debd444bb32929cd74e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:06:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9360fb60a80e48e9856065e6320231df\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91504dc494714a64a20e46b73375cfff : SUCCESS in 1h 27m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9774e14e84c549bba5482ff5e21ec099 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950fe1118d1c4f2ab96bed637a931c61 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c38e701cf54439b81c2ca3caba8ca79 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a02099334fd745449397629fcf7a2bcf : FAILURE in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb3327ee2d7549af8f4108237aef3c67 : FAILURE in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e51e5f8351aa4b068ac3f71ce186dda2 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7511c5be966242739694394221952e7a : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/728be65bb1e1474eafb5b8d95e910471 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73de989312ae4c1a8f3a570240e0422f : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab1362f16c0a4d8d8716af69dd967425 : SUCCESS in 7m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/784976e571c0480f8ffad0e1b6361d37 : SUCCESS in 5m 37s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/f45ced1135e045ac88e4f38912fbd57c : FAILURE in 25m 47s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/25bc0f6fc634436b909f3c8510c5d6ea : FAILURE in 20m 30s","accounts_in_message":[],"_revision_number":6},{"id":"c47d2f1f7ecdb382dbe0fffe74f004af83d9e2fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-21 11:04:07.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"3d7b2ba990ef725411b520f849e6acaa3b7af683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 13:06:31.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb5743f029b44ebd91c59c37d4733ae0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6785a6130dca4ae8a8716feac32deb15 : SUCCESS in 1h 16m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71271a0f3c54427bab65e0ebe84af2c3 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3bf09ab7c8747c3b36d4cdee9b63ec3 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58e1a4b8ec624650b8c51c4c95c40127 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7cfd0d157d64341a48b5f330331f3bf : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1db14bf7b0cf4f3db46b4b719cb2fe29 : FAILURE in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edd293ba8f174185af7357803ececf9e : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a3bb12274f6488c8dc1e6c8d79dba79 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7822cddd262f422da0f6cc97d7684bf9 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81104bdfc84848cab1af7e4d8c502550 : SUCCESS in 6m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df214360e5a34e3fa37af7aea47a3f11 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94a70ff6a5e04b74ba9385083932afc7 : SUCCESS in 5m 13s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/3ca03080ce4f486eb8d8f7290e0ef585 : FAILURE in 21m 06s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/5937834727e9438ba69c51a2bbc9555d : FAILURE in 18m 51s","accounts_in_message":[],"_revision_number":7},{"id":"4854aa4f80009949f4d49e1c33517123e6cc6e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-22 08:12:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"ce38e78cb30dbb19a06baebac300549c8eb8f7ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 09:56:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72c000f94e65433d90584332ade673e4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aca0cc5ffe8a4682b535834a5390ead2 : SUCCESS in 1m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15c15a76949142bd8f4798dc3c72f7f5 : SUCCESS in 1h 40m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde7c95a88e144879dbd1a8393fcd2b7 : SUCCESS in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4b7997578e4c7085808c5a4f88266a : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac170a242fa541baa40c59ec9a9fac60 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4cfd21770f24caf898949d8955ead6e : FAILURE in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f51264fec0742328bfcd217a102b9b5 : FAILURE in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f6f6ac118c146d997e0b90072cdff9a : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d317e4c3de994f8482adc263882a0775 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b1f87e1d5d4300acb150803d902a6e : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/471fee27ebeb48738df17217750c60fe : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56a5737bd1684672b72a12b30e670e3b : SUCCESS in 7m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5165a31327374257abbccb66fe9b92d3 : SUCCESS in 4m 38s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/a8e091e688cc437a9a1e807170732250 : FAILURE in 11m 47s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/fd66fb7b8b2844718121953d5368dacf : FAILURE in 18m 29s","accounts_in_message":[],"_revision_number":8},{"id":"a6a8de3b9f4702ef0a01b7a57715f59bfd1e8f7f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-25 08:57:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ccdb3ef30741f77765e51f49490e44bbbe6cc6a2","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-25 08:59:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3333c6764859f6fd6c1a97391065ba5762bffe91","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-25 10:01:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fb6f3f910c42f506a4241fe1433928f882f1c472","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-25 21:48:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bd7cd7371cd2d7f48e740e958daaee2edc26626e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-26 02:52:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c288cd289f973ba7e2d664c791c252d1a51d7772","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-26 10:39:20.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"26b4fd130c877c5b2a160e6fe60f045051f2a425","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-17 18:55:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0409e3ea3a40b0ea3ba3cac24bdaa841fa93e93c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-19 00:45:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a001505880a5507ac9e515b070367f52615219ba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-19 21:37:54.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e580d9bf9c5ee467430052d590adaff48de9ccdc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-04 07:50:28.000000000","message":"Abandoned\n\nreplaced by https://review.opendev.org/c/openstack/pbr/+/938030","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f0f06e605bbd2b7355243f1fa846cf8b13e51481","revisions":{"8208a38e062fb4b28459aee16feffdcbd904a509":{"kind":"REWORK","_number":1,"created":"2024-07-16 11:23:51.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/16/924216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/1"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 11:23:43.000000000","tz":120},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8208a38e062fb4b28459aee16feffdcbd904a509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8208a38e062fb4b28459aee16feffdcbd904a509"}]},"branch":"refs/heads/master"},"fbeae2094aefa41fb0ebbdc2339a8628d69e9e99":{"kind":"REWORK","_number":2,"created":"2024-11-18 17:00:24.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/16/924216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/2"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-18 17:00:06.000000000","tz":60},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fbeae2094aefa41fb0ebbdc2339a8628d69e9e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fbeae2094aefa41fb0ebbdc2339a8628d69e9e99"}]},"branch":"refs/heads/master"},"a3bddb03b5fe13598af5b0e52287a3ba4a63c131":{"kind":"REWORK","_number":3,"created":"2024-11-18 20:40:23.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/16/924216/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/3"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-18 20:40:05.000000000","tz":60},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a3bddb03b5fe13598af5b0e52287a3ba4a63c131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a3bddb03b5fe13598af5b0e52287a3ba4a63c131"}]},"branch":"refs/heads/master"},"d9ee3f1b7c2774b2b9a76117db34c4ecffbfa930":{"kind":"REWORK","_number":4,"created":"2024-11-19 09:38:40.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/16/924216/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/4"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-19 09:37:00.000000000","tz":60},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n6. remove install of cython3-dbg from installation job due to its\n   absence in ubuntu-nobble\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d9ee3f1b7c2774b2b9a76117db34c4ecffbfa930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d9ee3f1b7c2774b2b9a76117db34c4ecffbfa930"}]},"branch":"refs/heads/master"},"cb7eff997db0f1ce13dc5c72c727f8098acd696c":{"kind":"REWORK","_number":5,"created":"2024-11-20 13:05:43.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/16/924216/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/5"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-20 13:05:35.000000000","tz":60},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n6. remove install of cython3-dbg from installation job due to its\n   absence in ubuntu-nobble\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cb7eff997db0f1ce13dc5c72c727f8098acd696c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cb7eff997db0f1ce13dc5c72c727f8098acd696c"}]},"branch":"refs/heads/master"},"3895e4f8340324b9b8e49a52ae3c53ef9ad1716a":{"kind":"REWORK","_number":6,"created":"2024-11-20 14:36:17.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/16/924216/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/6"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-20 14:36:09.000000000","tz":60},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n6. remove install of cython3-dbg from installation job due to its\n   absence in ubuntu-nobble\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3895e4f8340324b9b8e49a52ae3c53ef9ad1716a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3895e4f8340324b9b8e49a52ae3c53ef9ad1716a"}]},"branch":"refs/heads/master"},"7b3df2bd8db8b3d546391060e858229c026705ff":{"kind":"REWORK","_number":7,"created":"2024-11-21 11:04:07.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/16/924216/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/7"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-21 11:03:13.000000000","tz":60},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n6. remove install of cython3-dbg from installation job due to its\n   absence in ubuntu-nobble\n7. add setuptools as docs requirements which now need to be explicit\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7b3df2bd8db8b3d546391060e858229c026705ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7b3df2bd8db8b3d546391060e858229c026705ff"}]},"branch":"refs/heads/master"},"f0f06e605bbd2b7355243f1fa846cf8b13e51481":{"kind":"REWORK","_number":8,"created":"2024-11-22 08:12:54.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/16/924216/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/924216/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/924216/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/924216/8"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-07-16 10:21:09.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-22 08:12:36.000000000","tz":60},"subject":"add python 3.12 support","message":"add python 3.12 support\n\n1. correctly add the setuptools dependency\nin test-requirements.txt\n2. move all usages from distutils commands\nto the appropriate setuptools equivalent, according to\nthese sources:\nhttps://setuptools.pypa.io/en/latest/deprecated/distutils-legacy.html#prefer-setuptools\nhttps://peps.python.org/pep-0632/#migration-advice\nhttps://github.com/pypa/setuptools/blob/e304e4dc00f588c9e247b4660c2fb0a194415be8/setuptools/errors.py#L9\n3. move logging to python std logging lib\n4. replace distutils.command.install in packaging.py with the setuptools\nequivalent, it\u0027s only used to run a command and should work(TM)\n5. replace usage of distutils.cmd with setuptools Command\n6. remove install of cython3-dbg from installation job due to its\n   absence in ubuntu-nobble\n7. add setuptools as docs requirements which now need to be explicit\n\nChange-Id: I3f4e9f95d10dbbba8499ac4354ff2bf402befd12\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f0f06e605bbd2b7355243f1fa846cf8b13e51481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f0f06e605bbd2b7355243f1fa846cf8b13e51481"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
