)]}'
{"id":"openstack%2Fpbr~924803","triplet_id":"openstack%2Fpbr~master~Ib2dcfb0f56af2410a8a6256fa77df664e373b2e8","project":"openstack/pbr","branch":"master","topic":"py314","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-07-22 12:42:34.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-22 09:11:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2025-07-22 12:42:34.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-07-22 12:42:34.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-22 12:42:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib2dcfb0f56af2410a8a6256fa77df664e373b2e8","subject":"Do not use the onerror parameter in shutil.rmtree()","status":"MERGED","created":"2024-07-24 00:01:27.000000000","updated":"2025-07-22 12:44:12.000000000","submitted":"2025-07-22 12:42:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924803-py314","meta_rev_id":"f5d9734b56b3f39be020a8b90430befbc4c36073","_number":924803,"virtual_id_number":924803,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-22 12:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-07-22 09:11:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-07-22 09:11:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-24 00:10:50.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2024-07-24 01:48:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-03 09:37:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-07-03 12:13:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-07-17 17:17:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"14fbeb6b9cbafeb2055d8fcc2fcc57b175449bea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-24 00:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efcefd965326e47e0bed24d4a127049ecc3b2071","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-24 00:10:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19141a115f4bb22dc7babf5ca22ebe2e8ab0f7a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 01:48:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/137c5397b6b74607a38711c03ce95321\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5053190e6b6f49819af4dbb65fd1eca5 : SUCCESS in 1h 45m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c65875ed1d4a7791d582d8cb9e2de4 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d196fd63f0d74ae68e9b0ee9bdb8c343 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/864c0a5d9732410493b32ec50c1af930 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/895baa3be4a241f8b916091a201a1939 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edce2cfeb0144ba3b492279a4457b999 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca58bb69645b4ae59d209d7f78c6aafe : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed803eee5be44889a17f45deb70d8433 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f38fe458cf8f4dc8892a48b25b8c78aa : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a36fe42b53784c0abc1e65713639b104 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d30ae0140c14d1c8eaba5ced89ac12c : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ef5d28c10ac430e93d65e9714a34f3b : SUCCESS in 6m 21s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/f3506aaa89fe49189f592b0cb359edc8 : SUCCESS in 20m 32s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/6b172398e2d140efac05feff363626a5 : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7158d1cf8567f1564556fe0909c2406c135926f0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-18 23:28:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f30816e1d6df943474d9f25bed7d6ddb92ed5ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 00:32:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b4d3428717d44c9a81b4fd0f64cffb3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14848b42a83648c985d5cbc5ecb0d97d : SUCCESS in 1h 01m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c00922ef5824a7dab1ad89d1eca4f3f : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e8a9ae48e14a1b993d038c5190fd79 : FAILURE in 2m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc74ca7c65724e6bae6941f976b71cde : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1366fac76344432398b7ea1068917a30 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00304ded1a1a410ca5b46471421a41a4 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cab02808981c4f06a173bd0dc95de023 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec226cc9068548868e779114478337a9 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b63183f6dd64632bfd44723f87dec2e : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f71c0b47fab48c0b3654f4b02f3d05c : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0ee299b5e734635b910ef5e5315d628 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/072389f96808441e8cbdcece36b6408c : SUCCESS in 6m 40s (non-voting)\n- pbr-installation-openstack https://zuul.opendev.org/t/openstack/build/7a3faa99939845c49c2d460f924fdfcb : FAILURE in 3m 39s\n- pbr-installation-openstack-pip-dev https://zuul.opendev.org/t/openstack/build/791bb98751134335b54d838d0158bf5e : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"db9aed45455ed2d1bbcbf771d9ff6fe859f2e5ab","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-26 00:13:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a0d9c0873723e77dcf095e638daf29de96291893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-06-26 00:24:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"561e80b5add7fc913d96283d7e369c1d75c59f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 01:49:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0a8eacc58e34d8caadcebc4d973b78f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/919919c9d16546a89e34eb988aaf319e : SUCCESS in 1h 23m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae5f7ce29d5493dbc740587ca63d273 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba4fbeffd1a43c5a5777f3f244faf23 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13ab6177a83248a4af940a93b0490ce6 : FAILURE in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35ff288de9d5479ca47c093c0db17129 : SUCCESS in 2m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0099440064d14796aef2368dff12dab6 : SUCCESS in 2m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e2e2a9355f54e05bf0e53e04dd70b95 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4903d3e248a4beb906932c59f72926d : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17434760640d4633b050b4280ea492df : FAILURE in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ff11882ba354b178fdd49b2a5dae644 : FAILURE in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/413008f7f3c441b591f83e7fcfa37fe0 : FAILURE in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b8c1b26403a494883e302a0812b7edd : FAILURE in 3m 16s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/60d74f18b4ea4594b398cd109b06321e : FAILURE in 12m 31s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d979e820758c4e3abdf2c60c796be415 : FAILURE in 16m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7c3cf0a91dcd4974a3474bc63630789a : FAILURE in 6m 30s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a7d123f117424f40944800b9233fba89 : FAILURE in 9m 15s","accounts_in_message":[],"_revision_number":2},{"id":"c4352867a17c72ef50cbd79cb104e4b54687627c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-02 14:54:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9b6706927916fbf69af3c0d4281e6fd8842c991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 16:47:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3d21b027bf041568157e6adbac24bb4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/401c5dfcb9a54267b2f9f21d5dd85744 : SUCCESS in 1h 43m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e37589852e455b9aac68deefa16db3 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43cd0df5d16540158a37d6325ddbedaf : SUCCESS in 3m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/285039e30f0f4640965d1e9f332f1e14 : FAILURE in 9m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/430a1328b4b841d9a39774b4d8f32aec : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ea5aededf6a480fb560343f0278ced6 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e59327ff8c6447da977106bb4709880 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e783a41a3a745fca64c4cf55ba9f6b5 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6f7d01a979a444c838b7aa6d5b02f24 : FAILURE in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9055ed51ce434b339ef6fadeaaf94771 : FAILURE in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50525ddab0184375aa86168db6615656 : FAILURE in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d43d53c667cf4d6f950eea3ace1f5b34 : FAILURE in 5m 13s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/b88884d9ca8345159e3701e9e619db61 : FAILURE in 23m 36s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/aa312826397b4fb8b2f8d169357af139 : FAILURE in 32m 51s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c204b6fcff9a44dda99638f10d069cdf : FAILURE in 19m 01s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c9845bf6f3d34b10956a65c31f8f95fa : FAILURE in 21m 46s","accounts_in_message":[],"_revision_number":2},{"id":"d8d7b15501449a97fad0f44fd969f390917266e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 09:37:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7d41d12f09511b928bb25569181af740047ef97c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 09:40:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"148185acc9ae0d5e16bda0b030af740ad83d6a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 11:42:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d12763af92947efba1eaf75c4cdd8e4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f169ae6ece6481cba66d8a5b97782c5 : SUCCESS in 1h 56m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bbf05a8c3544e16b05cd5668b2d092a : SUCCESS in 2m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ecf917cecc4055a71873e0ec71bcb2 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb323e4c08ea46ec8ea0660b96b84ba9 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cbae526186af42c19fca7415091a3fd7 : SUCCESS in 2m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0edd0f7636b4b568573b0d92956dfd7 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de81abcd6c3049d78c0ef9f8f7dda863 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a28eee8ee984f3da0b55942d98b89fe : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc80b520e9a64d419fca2019276cdab5 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/212a891a570b4c12b6030548b09a8b16 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be2b2e7693bb4eadac9966e8ef46b2b5 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b36410d1d1ab4d31a38bbd7548cb80b0 : SUCCESS in 8m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d9d06bac4ae47038359240a4bfaf088 : SUCCESS in 11m 26s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/64b2863fd6f7485bb03d5063e4f88646 : SUCCESS in 26m 30s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/f668737351d14d2cb11f6ad5cfb04967 : SUCCESS in 21m 16s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/fe1fb9483c6c4d29b18f08dbf89aaa53 : SUCCESS in 22m 53s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/2356e49ad1df4563b2768be6851b83d2 : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ac3bad7982fca6c08a99edaa9486000f824b4cc0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-07-03 12:13:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bab83a1fb8223d60ce444022490fdc9e71d809b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-16 10:09:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f87ff74e362475bdaf7a866e84f57414eea23433","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 10:10:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1381ecd739349c4e385b4fbda2e26ee3d9151ea0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 12:03:25.000000000","message":"Patch Set 3: Verified-2\n\n(1 comment)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23bdc720aa8d42b6b7cfe172d1d99f55\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b73708be05842a8991e0b96cd42f329 : SUCCESS in 1h 48m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484d361c26b8419f8f089c857fd85739 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a792538dae543ac9d2e3506987a585d : FAILURE in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e05624820d49f98caf320e0ba55904 : SUCCESS in 9m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1272e9dd642442ee987464d0d94a9c90 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b4f3d1844224da8be3c9a3d8efa26ac : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7141bc302c154bc889715a80349d563b : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ede2fa789ea481aba04d5cddf9ca4ad : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/796b18fd46804635b872e1f67df7dd3f : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8dd9f0a342bb4a799c8fe1359a5555e9 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/721db53925984906a627a32a928f0574 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61e406d3606a473eaaafea486dd8dd45 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e84041f0cc6c47efbf8bdb9c36af4100 : SUCCESS in 4m 07s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/697ba2a0effe47f19ab52ef1ad3c3718 : SUCCESS in 28m 21s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/5f63446561104e81a6318555c20c67f9 : FAILURE in 13m 52s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e5e6a2f14b35441295beb3b89829415d : SUCCESS in 36m 48s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4a664ac0912c40608c3add89840f03e5 : SUCCESS in 17m 03s","accounts_in_message":[],"_revision_number":3},{"id":"35455bfc7c65368e4129a4a85a9e211374040903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-16 15:49:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f340b2ed7dd0d75a1915ed6dac7db6a3e52a38fa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-16 15:49:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bdbe0f5e4de041ecd2b8b36bfd8c2c4aee01784a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-16 15:49:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3a9bfa5db6b46c40a74b0135abaa0c221fe57d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 16:44:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ca883654be4d5c95772ec60153a853\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e11080d589934ac599292e7f11fc7f53 : SUCCESS in 53m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7fff57019644eb8f9fb7c065420b57 : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3033ed73183e4abc8e418c77a12b09e2 : SUCCESS in 2m 35s\n- build-python-release https://zuul.opendev.org/t/openstack/build/457a4ee75e91437cab4d88bd1e488515 : SUCCESS in 1m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2c9cc3d7588470c857d233af03eb983 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89bd1e5392a440acbc5679b1bf68ca7f : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4e5e9b886dd406fbf96d7251171d49f : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d45e417d51e04abcb13da2e875f63499 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0303e2f625604a0cbc9ffa7b20bf6845 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bf384efa0bd40378fa77e22cec0288c : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b01ebffbcd240a5b6d0ad082281a0ec : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50d9850cea3c4999b704f405c643daee : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c7b6460b33b49b199061598e6d5860a : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7faa4b7854034ac097eb201050b7df83 : SUCCESS in 11m 35s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/d9d59247a40e47ff800abd55183d8d5a : SUCCESS in 27m 49s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/29b78d674ea54bf39835bd03052e8746 : SUCCESS in 36m 36s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/844a178db8684169975d65c86c268dec : SUCCESS in 43m 45s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b76aa686c9d54b9484a70b6ec2380cf7 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":4},{"id":"5c17e26b08258fb3b18545511bdc39d9fbaf338a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 17:38:48.000000000","message":"Patch Set 4: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/pbr for 924803,4","accounts_in_message":[],"_revision_number":4},{"id":"b2c66289e1c55ec7ebcb0bd8a39bd706037e0e18","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-17 17:17:17.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks like it needs a rebase.","accounts_in_message":[],"_revision_number":4},{"id":"c3fc0d78cc75c9ae81b33eca69d6b9d8fff9cbe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-22 09:11:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c67432ac0c8c8aa84f404b6389cf04829ac29424","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-22 09:11:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"71aea918ef2bc1ae6983e7a09563fee371372f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:13:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec4cff89872243218430c7ab93108cbd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/320a4d4ff0f04ff9b3c0e65026a82d07 : SUCCESS in 1h 48m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0efb491c54d4171884459d3de6acf8b : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe716c9d840549188e64cbee377d7936 : SUCCESS in 2m 22s\n- build-python-release https://zuul.opendev.org/t/openstack/build/4633792a52fe409bb87d3d92b770c764 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3940563b7b854f88b9980049ecb1a4fb : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15219ebb0032455cb42820e2af844ae3 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4434996cd4ca4f5a92279c639578fd2c : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d9be5934e8542fab1c0546e7638c2d9 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df274da2217845e5916daa2ea192a475 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a9d6a90574e4d83b38f83c80b79b2d6 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88ddd1f08851455cb19c20b02294d9ed : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/061a610afa8043928331a5433be19d80 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d09ae35c3c24d5e8241e1f308fabc31 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13fcf2528a914678b2140cba22b67122 : SUCCESS in 7m 39s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/c8e9843c01d9490b8d738437f0c5f204 : SUCCESS in 23m 06s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/083d9beabc104e7ebd52f4064661df33 : SUCCESS in 26m 00s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/3546ef33c05a418abc9cff773b59199d : SUCCESS in 34m 39s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/483c165741f146f9952483d8dc104ebf : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":5},{"id":"9f8e82478e93440e962534fa18c895a981366283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:15:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a97522f39eba12af8b6a92ad102184224aefc428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:42:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a08b1ea630e4d18a862e4eb4e4622bc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55dae90618be49ec978d5c78a5b2dda9 : SUCCESS in 58m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee5399ec05e4f80bf784e997de9d3e3 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95479a1baa1347ceb98dc97cf44efae1 : SUCCESS in 2m 46s\n- build-python-release https://zuul.opendev.org/t/openstack/build/86d6f77c3dc946a290ff6b3027b1140b : SUCCESS in 1m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4ca853886144032a4141c9187f1c6fa : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ada454b5cee244968150afe291637f28 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5042e3a69bc40769d8899a6db6a5986 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66fdbeb19d7b4489a6dee8d5c6583055 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94319d27c8a048e29e21dba8add0ffa4 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ed6e89dc9c43bea0d10df1d47df5d1 : SUCCESS in 8m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/153d35eef5734ebf8d33643bcc806c88 : SUCCESS in 8m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35829fbbe0924b619aa69e0ca9b3212a : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d6d13523f5b43ef918a1832110af3d7 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5ce455a7e4a464d88ab3509924ee40b : SUCCESS in 10m 11s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/854f9fa1ac5d45f2993fde2fc45686bc : SUCCESS in 26m 07s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6c302981206d40e697fe9e5dc95363ca : SUCCESS in 22m 41s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/37a245e93b564ecabb6a6caa2a5c48cb : SUCCESS in 34m 37s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ea50badf2bc842849bfbd9426436c91b : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":5},{"id":"fc84f7425253005848f67ffe26c61ece8ea3735d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:42:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"f5d9734b56b3f39be020a8b90430befbc4c36073","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:44:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36f475e85565419fa40d7132deecc128\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbba572a5fb4b92bcb1c92155ecdd48 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4a17dcb4800a0a7f3739228e211b4d8915c3440b","revisions":{"f3f1ff859e17506795900b46a201067ae3c33a05":{"kind":"REWORK","_number":1,"created":"2024-07-24 00:01:27.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/03/924803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/03/924803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/03/924803/1"}}},"commit":{"parents":[{"commit":"46ff9dd96718cdefe72a1a01447e14491917217b","subject":"Add openstack-tox-py312 as non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46ff9dd96718cdefe72a1a01447e14491917217b"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 23:57:17.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-24 00:00:39.000000000","tz":120},"subject":"Python 3.14: do not use the onerror parameter in shutil.rmtree()","message":"Python 3.14: do not use the onerror parameter in shutil.rmtree()\n\nThe onerror parameter has been deprecated in Python 3.12 and will be\nremoved in Python 3.14.\n\nChange-Id: Ib2dcfb0f56af2410a8a6256fa77df664e373b2e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f3f1ff859e17506795900b46a201067ae3c33a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f3f1ff859e17506795900b46a201067ae3c33a05"}]},"branch":"refs/heads/master"},"d5089489a7757bd597835892e1d9d1e884a8f07b":{"kind":"REWORK","_number":2,"created":"2025-06-26 00:24:46.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/03/924803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/03/924803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/03/924803/2"}}},"commit":{"parents":[{"commit":"5d4a1815afa920cf20e889be20617105446f7ce2","subject":"docs: Drop SetupTools from pyproject.toml example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d4a1815afa920cf20e889be20617105446f7ce2"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 23:57:17.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-06-26 00:15:42.000000000","tz":120},"subject":"Do not use the onerror parameter in shutil.rmtree()","message":"Do not use the onerror parameter in shutil.rmtree()\n\nThe onerror parameter has been deprecated in Python 3.12 and will be\nremoved in a future version of Python.\n\nChange-Id: Ib2dcfb0f56af2410a8a6256fa77df664e373b2e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d5089489a7757bd597835892e1d9d1e884a8f07b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d5089489a7757bd597835892e1d9d1e884a8f07b"}]},"branch":"refs/heads/master"},"537023d7f732f3d0ec4481e7e8fde3358087eee4":{"kind":"REWORK","_number":3,"created":"2025-07-03 09:40:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/03/924803/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/03/924803/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/03/924803/3"}}},"commit":{"parents":[{"commit":"5d4a1815afa920cf20e889be20617105446f7ce2","subject":"docs: Drop SetupTools from pyproject.toml example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d4a1815afa920cf20e889be20617105446f7ce2"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 23:57:17.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 09:40:20.000000000","tz":60},"subject":"Do not use the onerror parameter in shutil.rmtree()","message":"Do not use the onerror parameter in shutil.rmtree()\n\nThe onerror parameter has been deprecated in Python 3.12 and will be\nremoved in a future version of Python.\n\nChange-Id: Ib2dcfb0f56af2410a8a6256fa77df664e373b2e8\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/537023d7f732f3d0ec4481e7e8fde3358087eee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/537023d7f732f3d0ec4481e7e8fde3358087eee4"}]},"branch":"refs/heads/master"},"1d39ef26b720e793a756381c7aa66ac8d6e300d6":{"kind":"REWORK","_number":4,"created":"2025-07-16 15:49:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/03/924803/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/03/924803/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/03/924803/4"}}},"commit":{"parents":[{"commit":"4fafbadfdf61de5f292407ac451b79efa0610afd","subject":"Merge \"Use sysconfig for sitedir path in test_wsgi in py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4fafbadfdf61de5f292407ac451b79efa0610afd"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 23:57:17.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-16 15:49:09.000000000","tz":60},"subject":"Do not use the onerror parameter in shutil.rmtree()","message":"Do not use the onerror parameter in shutil.rmtree()\n\nThe onerror parameter has been deprecated in Python 3.12 and will be\nremoved in a future version of Python.\n\nChange-Id: Ib2dcfb0f56af2410a8a6256fa77df664e373b2e8\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1d39ef26b720e793a756381c7aa66ac8d6e300d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1d39ef26b720e793a756381c7aa66ac8d6e300d6"}]},"branch":"refs/heads/master"},"4a17dcb4800a0a7f3739228e211b4d8915c3440b":{"kind":"REWORK","_number":5,"created":"2025-07-22 09:11:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/03/924803/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/03/924803/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/03/924803/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/03/924803/5"}}},"commit":{"parents":[{"commit":"15d65487017e9d977958431b8a83c8d5e57b7c01","subject":"Merge \"Remove support for custom compilers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/15d65487017e9d977958431b8a83c8d5e57b7c01"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-23 23:57:17.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-22 09:10:59.000000000","tz":60},"subject":"Do not use the onerror parameter in shutil.rmtree()","message":"Do not use the onerror parameter in shutil.rmtree()\n\nThe onerror parameter has been deprecated in Python 3.12 and will be\nremoved in a future version of Python.\n\nChange-Id: Ib2dcfb0f56af2410a8a6256fa77df664e373b2e8\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a17dcb4800a0a7f3739228e211b4d8915c3440b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a17dcb4800a0a7f3739228e211b4d8915c3440b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
