)]}'
{"id":"openstack%2Fpbr~938332","triplet_id":"openstack%2Fpbr~master~I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad","project":"openstack/pbr","branch":"master","topic":"deps","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-01-06 16:00:15.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-01-03 21:30:00.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad","subject":"Set up PBR with setuptools as a dependency","status":"MERGED","created":"2024-12-30 22:44:58.000000000","updated":"2025-01-06 16:01:19.000000000","submitted":"2025-01-06 16:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938332-deps","meta_rev_id":"56fc58767109bf093381e4854c10a4981033ff62","_number":938332,"virtual_id_number":938332,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-06 16:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-01-06 14:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-03 21:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-01-06 14:11:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-31 01:05:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-03 19:50:15.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-01-06 14:11:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"4b87875fe0d03cd2491d5e9a45247799b939fc7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-30 22:44:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca821f472199ceb533a453bc06fa4a34d546c69c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-12-30 23:25:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3af9893d95454068c4f95b7514c6eabcaa426490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-31 01:05:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/842cc4d4de884537bf1861d804a05d08\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ce163c278bd41a591f5c507e95255c1 : SUCCESS in 1h 31m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d7e8981fef4a7eb13314b1cb71cdc8 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a765ea2fd7044bf18c2a3de4c33a26c3 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5664cec13f934b02b11be453fd51dab5 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8ca4796bc9e4ce1a45475d9b29cb01c : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b8d90c2764c4c6fbe34cdcff8fc3879 : FAILURE in 2m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25255c45adc14eb095c297fae3a3b5a7 : FAILURE in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/524bfbb999044ede8d102c54cb0018f3 : FAILURE in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33a6d87d05424f9eab375e889132e234 : FAILURE in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf179a26f667472290495ed3ebc1158d : FAILURE in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c86e6d22ef1442c79ddd411e1ead657c : FAILURE in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a158ddbe427b4f5281c199581c726efa : SUCCESS in 4m 30s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/382d4c18488e433a9d27507d23345dca : FAILURE in 6m 27s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/ab2959ac18b84d489c02748aa0708d5e : FAILURE in 5m 05s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/17a34fa05d124bca9777f130610a4972 : FAILURE in 24m 52s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0c935c2af4924ad4a2726021be0b0e10 : FAILURE in 24m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d96ba6aeb4e273bb257c5497468cef10ec289a77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 17:35:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e20ec41ab76f6be93ec7f92034b0071cda41d4b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 17:41:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2fff3322324cd17a2da385dea7fa393e305e0421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 18:03:22.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d1d960711e48ed479f7adace58c506ae0ebafeb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 18:58:16.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"0e4d42277bd721500654bbf6cb8036226b2ddc35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 19:49:24.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"769f8b72b4eeebe7b1aa8ea3b52c8aab4450d679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 20:18:59.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"883c237214a5d0d3498cd4654620a3156897bdec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 20:31:02.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"8e2bb2e706f4664dd7d735865d084c60b7d218d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 21:18:47.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"81c40949ecacd436884266fd38312ff97615b7c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 22:21:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"bac603a728a33ce96bd2cd8c47e1697b31ed0fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 22:28:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3d06281718d85a36346a6dfbb0d0b267c5d00cef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-02 22:58:40.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f8317c7d07edfe2095f2fb072442d199d35de4d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 00:08:54.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"17604fd5d51d11b2c08799cc224eef09f9cf3d86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 00:22:08.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"1433390de1c4e3de31313adbbe8063de1b46a73e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 01:21:46.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32d0a36f32fb4be8a8741a3adf5c861a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d59d1b57f0f43f09705aa276afed450 : SUCCESS in 57m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921dcdcb0d3649659b40eab83e0c523c : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbca7b03cd26412381fed22f4f93ad4f : SUCCESS in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95e547bf51c6488691cd112cf17a47f9 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba2b94e510e44ac4b8ced8fc39719785 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af15f19bf6974e6da6b845d39ed76e56 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32bbdb9b3d534570a3468a332e711b31 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42900acd15cd4edb82f7531e07478ae7 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b7b6102ab4b4585bb24f8209c124a7c : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0b51dd3eeee4bc4916e93e203b5acec : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81a36cc6cde241398d33f9301f87caeb : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5c936c96c3145debf52159a9369fe81 : SUCCESS in 4m 24s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/fb7989f75ed045d1a6033364c50e1dcf : FAILURE in 8m 24s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4be24896acc045f898ab07bb28b6f9ea : FAILURE in 7m 12s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0cb42bdb63824b4e86a1b012a825516a : FAILURE in 18m 46s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/be3dc27f4dde42b8adb68909e95eae0e : FAILURE in 19m 26s","accounts_in_message":[],"_revision_number":15},{"id":"1b1b89d2d950f2a4fe98d3979b51288095e10439","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 16:13:40.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"45ae61b8004b0d90b22d4c952ba7a9aee3cb3e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 16:49:03.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"5446920fc289b1be34a4b4f06fc7fd85de82e6b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 17:12:13.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"f5cf8c90e606d9130e8053022e6a2426c12ffe85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 18:35:15.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/867e21ce17df45b3bc44d549ce5e5941\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d652bd5a5ec6460cb86600ac5178a2b4 : SUCCESS in 1h 21m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6740bfb1a040b69be9085c4da183b9 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2231334cbb456c85881ab3e6640444 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f624a879eb574bb09ab94eb826779538 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64f71c80ae3540d488d8f0ff55c86eb3 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22de8b8170634a06a590eb8ed7164078 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9f8da3fc8b14905b9e9e0c8ca0843df : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6822ed9f04b4258854c8f89ba48feec : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e08f8a1b54964d98a9c0195380f1c364 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d2b0213fe1a4a27bccd9a8f0a1955ce : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be9471a575ea45759343330b8010c72e : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8844a2e1753143058fd1e1a64dabbe77 : SUCCESS in 5m 04s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/3259c1503db541449771fa9bae9ebf1e : SUCCESS in 22m 34s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2739711ea61d4de8b71e2c6b57c89f15 : SUCCESS in 25m 30s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/09fdd75c04634286a4d3450a8b62672a : FAILURE in 35m 09s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4619f520f7ee44eba889b588c7995dd1 : FAILURE in 38m 52s","accounts_in_message":[],"_revision_number":18},{"id":"2a9ca5e59616671534b1ca4b9f21551e66032c24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 18:36:16.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"528f81e99eacc7ce59d506d3f2ce94dd652e8ed7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-03 19:50:15.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"bfdf9afc1d7ad2764c1fd9de0c07456a3c6434cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 19:59:33.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fc3d74f5e8142c4a673579709daa813\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd711e251add4ca5a39e684c36f427b8 : SUCCESS in 1h 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8efb75e4d1fb41df9aeb5d0d0fdff8f1 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40064c98aad5434097a8956489d58338 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2f5c71ee69d465b967f640be123e7a0 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3548966f2da6409db9a844e58bd14136 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ee4c3b6c2d144178f042fbb5640a5b3 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/635134df91fd413eacf7a3c3f2e016fd : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55f8e34cf4d44b7398bdb09207feccdc : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2808016d18754b95ab4eca183395d326 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/913506e56b2d450eb5d365d5bf987ed8 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e46fc1496ee483a9112fcd0880c4292 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c130e375e94340b99abbaa19c8275171 : SUCCESS in 4m 09s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/cd79ec95896a46fea7701c0e0aa9e63b : SUCCESS in 16m 02s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/36a176c60cd24f80be39a5874567c324 : SUCCESS in 16m 44s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/de213ba4d7e54c738140fd94bead2f30 : SUCCESS in 35m 52s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b0e4c1e6a65044d7aad4f9fc19f58ffc : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":19},{"id":"a3749e4b7060acca00894a0daea56bfcf985ee08","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 21:02:58.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a05fbb71b44ef09245b79393a17ea90118860ca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 21:20:35.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"6920df774b178b68a6beb9d23964c6b33d672759","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-03 21:30:00.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)\n\nNote that I\u0027ve compared sdists and wheels of both the tip of master and this change (with all its parents), looking for missing/extra files or differences in metadata content, and have found no concerning differences between them.","accounts_in_message":[],"_revision_number":20},{"id":"58b27d07a2a75a77342a715d551d663ec44dc583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 22:46:47.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dee8fb016c474cdebba4e0c7c22f9eb6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c2c25858efa40a1923316305c60ecbf : SUCCESS in 1h 17m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33792a76fa9b41a5baaa1b806fb1cea2 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205f3b08e0d8426a8a0dcab406863f33 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82058f96d96c4f639bc50e17b0c41c17 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3884f8a77d64c0fa56b2102c9da16d3 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8ef55eb9eed4009a430c26c38cb11cd : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/518ccca6fcdb4b9ab8f46eb9eea989c9 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03364fd53d21417b9ac6c95bff7ba0b9 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c246b37059274cc1aa25dac4bf6eaf4d : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35820208ddbb45a38d5c493a4fb800f7 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88bfc86f0f6449d499e51bdcfb3db225 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3aba5cae552406e9053de6e924a6a95 : SUCCESS in 4m 34s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/9d78ff288568400e9c2134f634cfcfcb : SUCCESS in 26m 44s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/14ff53f476c443f89b0e385d1f1d16cf : SUCCESS in 24m 42s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/1e7e1e2c9853408eba13bd5c6431c3a8 : SUCCESS in 35m 45s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/dd3478bcd53841b8a6d8517a427d72ff : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":20},{"id":"1b4cc89f8da270c400a5757f4f0cf530c69e28e2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-06 14:11:50.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"67e3ad15a3c4bb8f42b2a167c8abba4d2a2e3098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:13:26.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"57c08fdcf6947050bc11d19b79af670e258d71f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 16:00:15.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b24f2625ff2646a38fc4b715b9d30c06\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/865600c9d0514bf0a8aaecaa031756a5 : SUCCESS in 1h 41m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa62dcf1e3924a7d8963aae53fc3d4f1 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43afdbe0b39461195bf93f449e4872b : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/830f3efcb10d400eb56c2fabf6547a2c : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d1127a662894d948c012ab1be48d90f : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3102db32a1254bd18a150515d258ff86 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65c9efaee5e244ada39ebb45b1666903 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/651845f04cd14282bb7cd653c4c96f2e : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b1b2e7ecc441f2b78ee9ea404fab16 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6248a6989e848cd9c7c043ff1c30b01 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c04618ffb9ad40c3affb180cbb47fff8 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1be9549ee539408ab2bf6f568d3f8319 : SUCCESS in 3m 10s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/455dbd04b69149c78ad7929f35097ee3 : SUCCESS in 24m 16s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/19421ca18043433184072a028141f958 : SUCCESS in 24m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9c5440adf8fb402eb3c09825a9672a4a : SUCCESS in 26m 57s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/baa6f7b053894eaab980401e652c0923 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":20},{"id":"ec5ce19bf5aaff901b092dc9e3a77d8d39dc729b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 16:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"56fc58767109bf093381e4854c10a4981033ff62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 16:01:19.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be8cb67abeaf47678537b24be9e73a42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f4d3bb169b4aee88f70658a443f92a : SUCCESS in 34s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"7cd74959d4a4857afd72f05812d4f7eb8fc449d4","revisions":{"1c22f5bab99b0180f3f30298b2d9ed5b0a31e458":{"kind":"REWORK","_number":1,"created":"2024-12-30 22:44:58.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/1"}}},"commit":{"parents":[{"commit":"9f1c57f0f73b78d04dcacace684ada54831affe2","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9f1c57f0f73b78d04dcacace684ada54831affe2"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1c22f5bab99b0180f3f30298b2d9ed5b0a31e458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1c22f5bab99b0180f3f30298b2d9ed5b0a31e458"}]},"branch":"refs/heads/master"},"11b1b96e1034919ebf241bb7900f157bd1de5001":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-30 23:25:03.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/2"}}},"commit":{"parents":[{"commit":"bdb99f96707bc23f1524352607e623c13cdffa86","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bdb99f96707bc23f1524352607e623c13cdffa86"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 23:12:17.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/11b1b96e1034919ebf241bb7900f157bd1de5001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/11b1b96e1034919ebf241bb7900f157bd1de5001"}]},"branch":"refs/heads/master"},"cd92728ff4109dc57c90934ed53269e69d49b1b4":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-01-02 17:35:27.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/3"}}},"commit":{"parents":[{"commit":"c19003005e8aa5d75d2146486c489ca459cc41a8","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c19003005e8aa5d75d2146486c489ca459cc41a8"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 17:33:04.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd92728ff4109dc57c90934ed53269e69d49b1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd92728ff4109dc57c90934ed53269e69d49b1b4"}]},"branch":"refs/heads/master"},"8737df938b6a241f420d46eff44e8a2fd0d581a6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-02 17:41:54.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/4"}}},"commit":{"parents":[{"commit":"64c8686d2342135dce732aaa47c4c500ccbc2a5e","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/64c8686d2342135dce732aaa47c4c500ccbc2a5e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 17:41:47.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8737df938b6a241f420d46eff44e8a2fd0d581a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8737df938b6a241f420d46eff44e8a2fd0d581a6"}]},"branch":"refs/heads/master"},"8d5cf77904f9fa0d5bc599ebcb1aeacb6f7b76ce":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-01-02 18:03:22.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/5"}}},"commit":{"parents":[{"commit":"dd1a841c864dcfffc747e1f7d1dad84fb334d476","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dd1a841c864dcfffc747e1f7d1dad84fb334d476"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 18:02:41.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d5cf77904f9fa0d5bc599ebcb1aeacb6f7b76ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d5cf77904f9fa0d5bc599ebcb1aeacb6f7b76ce"}]},"branch":"refs/heads/master"},"218812aeb754817ffafc6100f091855bdb021f30":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-01-02 18:58:16.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/6"}}},"commit":{"parents":[{"commit":"8fc6027fd99db960f923a0039088f6b82be3b4b1","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8fc6027fd99db960f923a0039088f6b82be3b4b1"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 18:58:05.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/218812aeb754817ffafc6100f091855bdb021f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/218812aeb754817ffafc6100f091855bdb021f30"}]},"branch":"refs/heads/master"},"f5a0df789cdd123b9bd5e88cd8a684e8cd68085b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-01-02 19:49:24.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/7"}}},"commit":{"parents":[{"commit":"8c0bd802eca83b65df94ed5fdcd72f0d9ae3edd4","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8c0bd802eca83b65df94ed5fdcd72f0d9ae3edd4"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 19:48:57.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f5a0df789cdd123b9bd5e88cd8a684e8cd68085b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f5a0df789cdd123b9bd5e88cd8a684e8cd68085b"}]},"branch":"refs/heads/master"},"8a2083092b92b34d4edef51ddf88d86758746e80":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-01-02 20:18:59.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/8"}}},"commit":{"parents":[{"commit":"67653ad73fed1ed2b89d9cf16798f47830da2e7e","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/67653ad73fed1ed2b89d9cf16798f47830da2e7e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 20:18:36.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8a2083092b92b34d4edef51ddf88d86758746e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8a2083092b92b34d4edef51ddf88d86758746e80"}]},"branch":"refs/heads/master"},"3cff873e8f619bcc77fe1e3a929138cad47a80ab":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-01-02 20:31:02.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/9"}}},"commit":{"parents":[{"commit":"11fde828b27b30645e2186936d7151dd36b57cd5","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/11fde828b27b30645e2186936d7151dd36b57cd5"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 20:30:56.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3cff873e8f619bcc77fe1e3a929138cad47a80ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3cff873e8f619bcc77fe1e3a929138cad47a80ab"}]},"branch":"refs/heads/master"},"963795ca7d17cc30f5e48fe8babb35fb11691d8c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-01-02 21:18:47.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/10"}}},"commit":{"parents":[{"commit":"526eb42d730f2f758e824a07c050dcd413f6f952","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/526eb42d730f2f758e824a07c050dcd413f6f952"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 21:18:30.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/963795ca7d17cc30f5e48fe8babb35fb11691d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/963795ca7d17cc30f5e48fe8babb35fb11691d8c"}]},"branch":"refs/heads/master"},"1503230b84900e8d8330c903c085860add375a53":{"kind":"REWORK","_number":11,"created":"2025-01-02 22:21:19.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/11"}}},"commit":{"parents":[{"commit":"526eb42d730f2f758e824a07c050dcd413f6f952","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/526eb42d730f2f758e824a07c050dcd413f6f952"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 22:21:04.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1503230b84900e8d8330c903c085860add375a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1503230b84900e8d8330c903c085860add375a53"}]},"branch":"refs/heads/master"},"f9a62c155534fc706f34633283ae6d187626a636":{"kind":"REWORK","_number":12,"created":"2025-01-02 22:28:43.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/12"}}},"commit":{"parents":[{"commit":"526eb42d730f2f758e824a07c050dcd413f6f952","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/526eb42d730f2f758e824a07c050dcd413f6f952"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 22:28:37.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f9a62c155534fc706f34633283ae6d187626a636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f9a62c155534fc706f34633283ae6d187626a636"}]},"branch":"refs/heads/master"},"0e14191d4ef36569fb327895d1722e2f868d01c8":{"kind":"REWORK","_number":13,"created":"2025-01-02 22:58:40.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/13"}}},"commit":{"parents":[{"commit":"526eb42d730f2f758e824a07c050dcd413f6f952","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/526eb42d730f2f758e824a07c050dcd413f6f952"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-02 22:58:23.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0e14191d4ef36569fb327895d1722e2f868d01c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0e14191d4ef36569fb327895d1722e2f868d01c8"}]},"branch":"refs/heads/master"},"7629c17abb242ee0ee4253e5fad4f6613ab47a43":{"kind":"REWORK","_number":14,"created":"2025-01-03 00:08:54.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/14"}}},"commit":{"parents":[{"commit":"526eb42d730f2f758e824a07c050dcd413f6f952","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/526eb42d730f2f758e824a07c050dcd413f6f952"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 00:08:47.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7629c17abb242ee0ee4253e5fad4f6613ab47a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7629c17abb242ee0ee4253e5fad4f6613ab47a43"}]},"branch":"refs/heads/master"},"cba51a53241414485a5edcf432808e9e4adbb407":{"kind":"REWORK","_number":15,"created":"2025-01-03 00:22:08.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/15"}}},"commit":{"parents":[{"commit":"526eb42d730f2f758e824a07c050dcd413f6f952","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/526eb42d730f2f758e824a07c050dcd413f6f952"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 00:22:02.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cba51a53241414485a5edcf432808e9e4adbb407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cba51a53241414485a5edcf432808e9e4adbb407"}]},"branch":"refs/heads/master"},"22dc6a94db6f6683231e1b69545d44376182ec99":{"kind":"REWORK","_number":16,"created":"2025-01-03 16:13:40.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/16"}}},"commit":{"parents":[{"commit":"ef56802eb8f0eec7013798ee03eda1751da9f0ff","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ef56802eb8f0eec7013798ee03eda1751da9f0ff"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 16:11:16.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/22dc6a94db6f6683231e1b69545d44376182ec99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/22dc6a94db6f6683231e1b69545d44376182ec99"}]},"branch":"refs/heads/master"},"987cf1065caa829a7beb1b0177623d5e31092b1b":{"kind":"REWORK","_number":17,"created":"2025-01-03 16:49:03.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/17"}}},"commit":{"parents":[{"commit":"ef56802eb8f0eec7013798ee03eda1751da9f0ff","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ef56802eb8f0eec7013798ee03eda1751da9f0ff"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 16:47:56.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nWe modify the integration tests to expand the coverage of our test\nproject installation as a result. In particular we want to ensure this\nforward looking set of behaviors works as expected while still checking\nexisting behaviors continue to work for packages that haven\u0027t updated\nyet.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/987cf1065caa829a7beb1b0177623d5e31092b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/987cf1065caa829a7beb1b0177623d5e31092b1b"}]},"branch":"refs/heads/master"},"f421bc93f49e60fe8616931d04298b9903892441":{"kind":"REWORK","_number":18,"created":"2025-01-03 17:12:13.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/18"}}},"commit":{"parents":[{"commit":"52ff6c86970f4754a7a0a688355cf33a0315086c","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52ff6c86970f4754a7a0a688355cf33a0315086c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 17:11:05.000000000","tz":-480},"subject":"Set up PBR with setuptools and six as dependencies","message":"Set up PBR with setuptools and six as dependencies\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress these runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nWe modify the integration tests to expand the coverage of our test\nproject installation as a result. In particular we want to ensure this\nforward looking set of behaviors works as expected while still checking\nexisting behaviors continue to work for packages that haven\u0027t updated\nyet.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f421bc93f49e60fe8616931d04298b9903892441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f421bc93f49e60fe8616931d04298b9903892441"}]},"branch":"refs/heads/master"},"6dcd0bc7275b444ced68127342c687970868c98d":{"kind":"REWORK","_number":19,"created":"2025-01-03 18:36:16.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/19"}}},"commit":{"parents":[{"commit":"0ad8a5a1ef9ff2e73c9cec224635bc3474700432","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0ad8a5a1ef9ff2e73c9cec224635bc3474700432"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 18:15:21.000000000","tz":-480},"subject":"Set up PBR with setuptools as a dependency","message":"Set up PBR with setuptools as a dependency\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress this runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nWe modify the integration tests to expand the coverage of our test\nproject installation as a result. In particular we want to ensure this\nforward looking set of behaviors works as expected while still checking\nexisting behaviors continue to work for packages that haven\u0027t updated\nyet.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6dcd0bc7275b444ced68127342c687970868c98d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6dcd0bc7275b444ced68127342c687970868c98d"}]},"branch":"refs/heads/master"},"7cd74959d4a4857afd72f05812d4f7eb8fc449d4":{"kind":"REWORK","_number":20,"created":"2025-01-03 21:20:35.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/32/938332/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/32/938332/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/32/938332/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/32/938332/20"}}},"commit":{"parents":[{"commit":"8f8d736a7f5057ee9dfad54e065c907713dba515","subject":"Use pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8f8d736a7f5057ee9dfad54e065c907713dba515"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-12-30 22:43:40.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 21:20:23.000000000","tz":-480},"subject":"Set up PBR with setuptools as a dependency","message":"Set up PBR with setuptools as a dependency\n\nFor python3.12 and newer setuptools is no longer present by default.\nExpress this runtime dependencies for PBR via a normal\nrequirements.txt. The expectation is that anyone installing a package\nwith PBR as the build tool will use a pyproject.toml file and if using\nPBR at runtime will list PBR as a runtime dependency which should pull\nin setuptools for the runtime dep.\n\nWe modify the integration tests to expand the coverage of our test\nproject installation as a result. In particular we want to ensure this\nforward looking set of behaviors works as expected while still checking\nexisting behaviors continue to work for packages that haven\u0027t updated\nyet.\n\nDepends-On: https://review.opendev.org/c/openstack/pbr/+/938076\nChange-Id: I7ef7472ae48c65a47581694f2f1fcc4edbb1d6ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7cd74959d4a4857afd72f05812d4f7eb8fc449d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7cd74959d4a4857afd72f05812d4f7eb8fc449d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
