)]}'
{"id":"openstack%2Fpbr~938410","triplet_id":"openstack%2Fpbr~master~I39947d1e37e07f4d7d651ec4102a58e141132257","project":"openstack/pbr","branch":"master","topic":"deps","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-01-06 15:56:49.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-01-03 21:21:57.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I39947d1e37e07f4d7d651ec4102a58e141132257","subject":"Test pip and setup.py install behavior in PBR testing","status":"MERGED","created":"2025-01-03 15:42:56.000000000","updated":"2025-01-06 15:56:49.000000000","submitted":"2025-01-06 15:56:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938410-deps","meta_rev_id":"6bbcb506919386d19c67775c83134fd64d3f1f2e","_number":938410,"virtual_id_number":938410,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-06 15:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-01-06 14:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-03 21:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-01-06 14:09:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-03 17:03:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-03 19:17:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-01-06 14:09:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"99f61207688a9c409ac856e869b54b20cc240114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 15:42:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fab02f652bf225ea128d1076cad73d90b1d7de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 16:01:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93c8fd0884442d14cbbdd0dc330519291535b22a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 17:03:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0e4e3396c3e4ae4b86c0cdb34c00149\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94b5de5bf5a0416c90777b5f4bba4760 : SUCCESS in 59m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c1afb678454bf09a21c721f5374174 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c9606ba65542e1a00da4692da2d710 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc3f030b1b064b75af4c3b26f2eaa31a : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/456295749547426d82a94208cea28e14 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ce796f82b694fa5b9a8062ae4d9921c : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72ebbb8da6d74eb7ba8e72c07af31802 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a6e567982d44ef3a155db9ddf4b03b0 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e547d7548d14f8eb9a6f2a023b6b3f7 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ccfd471390347d48d68a30a1025b790 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c68ed2edaf224e29a82c3124121c9d21 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9da9970eab154f82b332a27f41279278 : SUCCESS in 4m 58s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/86707a1da3f341f88f358d1a4201838e : SUCCESS in 23m 39s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6e5325e84fed457ba8066e208e9eba62 : SUCCESS in 15m 41s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/af4dfd157faf4710be735f8bb289c03b : SUCCESS in 24m 41s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/08484886edb84b8785533932781a8a97 : SUCCESS in 23m 37s","accounts_in_message":[],"_revision_number":2},{"id":"0f091a73611e9d60eb3d7a251e6c86175b23c130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 17:12:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b4af7e6b8e18f97782fa7473ff13362797688674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 18:28:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/779fc563ff2d462390ef5c01a79a818d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdac525f2c204c18890ea079f73deb3f : SUCCESS in 1h 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3b5770b1c9941808223f482ba5de07a : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3d8526c93704316b17a5242bb17e050 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/935795746c7a41d7b8118e2c3697adfd : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e87c0bf1b1f54fd899d108c344ac773d : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6695896d66f148b89d4a305a49ba1a0c : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e481205e7be408db1a68e64bc13bc3e : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602b35acd8894040b885c537c858db30 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b41b71f56f604d7c92e4d96fee734411 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c8d4ee5a68944e79b0b433cc8f52780 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/108d139f00994feca789de3c85bc957e : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5af0169821134e7cac1c21028eb754ce : SUCCESS in 5m 35s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ea6d7c158723430186956874a2e73543 : SUCCESS in 25m 56s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/288ccf8d794b4688ae48b44329484445 : SUCCESS in 22m 45s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/64dfa9c8e25141f9abb74c6c5363c919 : SUCCESS in 29m 56s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/dc2d13f8583a4686a2d55538e15276cb : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":3},{"id":"8ede720a34d9e2a80d65397ce4e1016d62473ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 18:36:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2cb6772a96d3c80d0ab2f68e77fda6e500bf1028","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-03 19:17:49.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)\n\nAdditional tools worth testing are https://pypi.org/p/build and https://pypi.org/p/wheel since we may get subtly different behaviors or lower-level errors reported.","accounts_in_message":[],"_revision_number":4},{"id":"9cf3da34f346bf2881cd3e81e63d34f997c4e874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 20:02:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e08d4b9b8e624ff88e5e8b1e413a4490\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/233e6304d03e4cfca7841b177a120db7 : SUCCESS in 1h 24m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d60eea2de18b47abb0000f29bfc803e6 : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f83ec37cf1e4a1f91e9acebcdec5815 : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68ee34cb99a1436790512ba78605014b : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32d3f9d63da648d5828190335245099a : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f953aeaef44a4f3083d776225f7c14bb : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f410d7a2ef02425dbf6f26d3980b4dae : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ea52948af8c4c1fbd225265f016d686 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4352ae76957410d8b20b3cc17883ac9 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d360a431ce6e492da3468b7338adaef7 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47d263a542d943ff80dc6344088364bc : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/488c569dd4204859b19537ef11c43581 : SUCCESS in 6m 08s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/57ed428e05a14090b0a6cd7b0ff0bdb0 : SUCCESS in 19m 47s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/59e8eb33e7094022941252b6fd130d1a : SUCCESS in 21m 02s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9430f4f8dfda47239027fac000793f5f : SUCCESS in 33m 20s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/65ec3dbbae534993b161c5aa6120c7e1 : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":4},{"id":"8e23c9c71108a8998a723fbc5b1d814253152d38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-03 21:20:35.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1e9ff23a26d96dbf9719cfd8389dbf1085febaac","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-03 21:21:57.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"daf8ad884079619f7136986fc1b1522498054008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-03 22:45:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a23850dce90a40dba883219743994b09\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5a0a1118d414ef48f43acbd7743f461 : SUCCESS in 1h 22m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b9616986254bdda9bf9a9ead4e479b : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5602370fe42146a1b4aebc95713634ac : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec96ba9c1448495aaa5b21c85908f50e : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6aa27988c16c41b8b460a2e09d5644f7 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acc991bd78e440feaef387fb3271e0b6 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/386716a6d9464d92878545b741f0d8d9 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/703bb0b117fc498cbcdbbb30d2d17002 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/255eb24f92c248cd8486338a825d7813 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/679665ff379a4ec3bccfa80cd35c8931 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ec66cf9fc724f9682885447a54846d9 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f2d24aac39f4ec281c4eb9c810c3197 : SUCCESS in 3m 46s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/28cbcc5443464eb4a94dcd7450f28340 : SUCCESS in 16m 08s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b79d918ec134443a9d257ca62c585e20 : SUCCESS in 14m 22s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/8d7d17c903d54fb39dc79e1e8adc137d : SUCCESS in 20m 26s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a6f794b0a0a846fa8c52d5fe8ddfae9e : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":5},{"id":"35f8fb5646e69ebf96de922538cfcd424b9bbb44","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-06 14:09:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"363dae20e11dd2faa1dcb304e373a1809f3bf74b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:11:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9c59a5bde2c91c4a89d213acbe32f6edc9b16cf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 15:56:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/422d96974ab34c20a3be7f9682021538\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9b903f8eea74d78ba60b150caa399a7 : SUCCESS in 1h 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ad5b4d15b44075979768468a33b2eb : SUCCESS in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b004c121f6247448ce04d245d1d1254 : SUCCESS in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d2e5b34f54544c48a3ce029fe9fd04a : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/001ea57e3f0b44da890afa77ca6b1f27 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba930cb144e4492d9d994e82fff84b38 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/115027c48b8e4af29be4d1ad639a384a : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7750f4e7302f489197d4242e164bec37 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b3bd4f062c04540b102fb5ed432fa4d : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed8e85e2e70747168cbc2ddea6a02166 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5f8b68fb1704bb78bc37ed7c62ff506 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c433bdcbe68443ea840bd0e4c24751b : SUCCESS in 4m 45s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/4f68fadce7784bdeb887594621a1b17a : SUCCESS in 23m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4e8bd4a7c04e41fa9c0925175354fb3e : SUCCESS in 23m 57s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/d51e645207c144b180e3d7fac130d4b1 : SUCCESS in 33m 46s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/5de3b5cdc2234262a7d723d50fb2a17e : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":5},{"id":"6bbcb506919386d19c67775c83134fd64d3f1f2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 15:56:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e1124b797770d32c684b7aa1812864f5671f094e","revisions":{"44cf4e611eacd7b48f8252605bd5c83d395d1c6a":{"kind":"REWORK","_number":1,"created":"2025-01-03 15:42:56.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/10/938410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/10/938410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/10/938410/1"}}},"commit":{"parents":[{"commit":"f797fc1c58bc1667fa1db618dca6f203b8ccc680","subject":"Update PBR testing for Noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f797fc1c58bc1667fa1db618dca6f203b8ccc680"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 15:40:17.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 15:40:17.000000000","tz":-480},"subject":"Test pip and setup.py install behavior in PBR testing","message":"Test pip and setup.py install behavior in PBR testing\n\nSetup.py is deprecated as a direct installation tool. Instead you should\nbe installing (or building) python packages with other tools like pip\nand build. To cover these cases add a pip install path to integration\ntesting after our setup.py testing.\n\nChange-Id: I39947d1e37e07f4d7d651ec4102a58e141132257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/44cf4e611eacd7b48f8252605bd5c83d395d1c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/44cf4e611eacd7b48f8252605bd5c83d395d1c6a"}]},"branch":"refs/heads/master"},"c5f0560f8ed9e2c6397a4f05e6d60c5afbb3297a":{"kind":"REWORK","_number":2,"created":"2025-01-03 16:01:53.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/10/938410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/10/938410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/10/938410/2"}}},"commit":{"parents":[{"commit":"f797fc1c58bc1667fa1db618dca6f203b8ccc680","subject":"Update PBR testing for Noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f797fc1c58bc1667fa1db618dca6f203b8ccc680"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 15:40:17.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 16:01:38.000000000","tz":-480},"subject":"Test pip and setup.py install behavior in PBR testing","message":"Test pip and setup.py install behavior in PBR testing\n\nSetup.py is deprecated as a direct installation tool. Instead you should\nbe installing (or building) python packages with other tools like pip\nand build. To cover these cases add a pip install path to integration\ntesting after our setup.py testing.\n\nChange-Id: I39947d1e37e07f4d7d651ec4102a58e141132257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c5f0560f8ed9e2c6397a4f05e6d60c5afbb3297a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c5f0560f8ed9e2c6397a4f05e6d60c5afbb3297a"}]},"branch":"refs/heads/master"},"6920222727ad7db6363a5053975dd6fd8d549b60":{"kind":"REWORK","_number":3,"created":"2025-01-03 17:12:13.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/10/938410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/10/938410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/10/938410/3"}}},"commit":{"parents":[{"commit":"f797fc1c58bc1667fa1db618dca6f203b8ccc680","subject":"Update PBR testing for Noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f797fc1c58bc1667fa1db618dca6f203b8ccc680"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 15:40:17.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 17:10:11.000000000","tz":-480},"subject":"Test pip and setup.py install behavior in PBR testing","message":"Test pip and setup.py install behavior in PBR testing\n\nSetup.py is deprecated as a direct installation tool. Instead you should\nbe installing (or building) python packages with other tools like pip\nand build. To cover these cases add a pip install path to integration\ntesting after our setup.py testing.\n\nChange-Id: I39947d1e37e07f4d7d651ec4102a58e141132257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6920222727ad7db6363a5053975dd6fd8d549b60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6920222727ad7db6363a5053975dd6fd8d549b60"}]},"branch":"refs/heads/master"},"b4567f7849ad03f5886d5c09a2f455055b939378":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-03 18:36:16.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/10/938410/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/10/938410/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/10/938410/4"}}},"commit":{"parents":[{"commit":"50fceaca0d2c393dbf2341db13a5fb3faf24a4f4","subject":"Remove the use of six from PBR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/50fceaca0d2c393dbf2341db13a5fb3faf24a4f4"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 15:40:17.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 18:09:23.000000000","tz":-480},"subject":"Test pip and setup.py install behavior in PBR testing","message":"Test pip and setup.py install behavior in PBR testing\n\nSetup.py is deprecated as a direct installation tool. Instead you should\nbe installing (or building) python packages with other tools like pip\nand build. To cover these cases add a pip install path to integration\ntesting after our setup.py testing.\n\nChange-Id: I39947d1e37e07f4d7d651ec4102a58e141132257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b4567f7849ad03f5886d5c09a2f455055b939378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b4567f7849ad03f5886d5c09a2f455055b939378"}]},"branch":"refs/heads/master"},"e1124b797770d32c684b7aa1812864f5671f094e":{"kind":"REWORK","_number":5,"created":"2025-01-03 21:20:35.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/10/938410/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/10/938410/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/10/938410/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/10/938410/5"}}},"commit":{"parents":[{"commit":"50fceaca0d2c393dbf2341db13a5fb3faf24a4f4","subject":"Remove the use of six from PBR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/50fceaca0d2c393dbf2341db13a5fb3faf24a4f4"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 15:40:17.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-03 21:05:10.000000000","tz":-480},"subject":"Test pip and setup.py install behavior in PBR testing","message":"Test pip and setup.py install behavior in PBR testing\n\nSetup.py is deprecated as a direct installation tool. Instead you should\nbe installing (or building) python packages with other tools like pip\nand build. To cover these cases add a pip install path to integration\ntesting after our setup.py testing.\n\nChange-Id: I39947d1e37e07f4d7d651ec4102a58e141132257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e1124b797770d32c684b7aa1812864f5671f094e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e1124b797770d32c684b7aa1812864f5671f094e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
