)]}'
{"id":"openstack%2Fpbr~938525","triplet_id":"openstack%2Fpbr~master~Ibe1a8987d8f3752b9ab7b0991e00893e1cd840a2","project":"openstack/pbr","branch":"master","topic":"run-release-job","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-07-16 11:14:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe1a8987d8f3752b9ab7b0991e00893e1cd840a2","subject":"Run python release package build job in check and gate","status":"MERGED","created":"2025-01-07 00:04:53.000000000","updated":"2025-07-16 11:15:52.000000000","submitted":"2025-07-16 11:14:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938525-run-release-job","meta_rev_id":"6611f69f049d314390566fef6ae5446db54b2546","_number":938525,"virtual_id_number":938525,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-16 11:14:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2025-01-08 18:28:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-07-16 10:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-07 14:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-07-16 10:09:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 01:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 14:02:43.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-01-08 18:28:24.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-07-16 10:09:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"1cf30e3de4050687b592b8b520f1760ce75ac220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-07 00:04:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"930a202e518e53b280b8002760d8043fc4e003ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 01:42:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab11e5ba6471448e810bb6d4edfa2d5b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0df4c911f134686b7a51f1cacafb83e : SUCCESS in 1h 36m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32661a06fed244f2964413c63dc69be6 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd1ae0ddd0841edbc69d14d4803db9d : SUCCESS in 2m 38s\n- build-python-release https://zuul.opendev.org/t/openstack/build/a46dd9ae2aa541bf90a1593bd0fb83fc : SUCCESS in 1m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97d917ddfeb14c4b8b1b804600a0166c : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e22fe86109e94731a8cb8e3e96829857 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/587b5e82233b40d18312e345c0bb4c39 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a301a7c11059466a9bbd2dd7ea8a2f97 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c480349c7724a169b3a6e4feacccb91 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bd900c9886a40f3aca063a52ce367c8 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e703520d1c044e02a4db36bad3be75eb : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b3a307e9abe4730ad3c195ba711d64e : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5f9f93ed37f4c928bfb88a7c473e04a : SUCCESS in 5m 01s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/03626375245442ce8759b69d2c7165a2 : SUCCESS in 27m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6433c2e67ab24f069d498808d441d5da : SUCCESS in 35m 27s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9391afb93616417bb0054979de235c47 : SUCCESS in 32m 57s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/2bf2e537b19743cf846527bed9f1a86e : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9248c06992f53a8d1d4e075a02d8b3728cc30ee4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-07 14:02:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c0bcb75fbe64f48029933f98897443c79fbd9f54","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-08 18:28:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e0bbd5c4badc463639eedf18a50b4e062a342585","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-16 10:09:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"526b8157a36b846b245e754a1865dd12979492b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 10:11:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d68bc1295a220514b89380d14af1bfb07ea92ae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 11:14:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b182a1b70074d35a4e6755a3d8e778d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c388bd9fce494d7f8bde2748c4764db5 : SUCCESS in 54m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17745b93da1446e993321c59ee95c912 : SUCCESS in 1m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e6dff3e37974e95914bf266a8086f56 : SUCCESS in 1m 35s\n- build-python-release https://zuul.opendev.org/t/openstack/build/8a62695bcb6a49cda8ecd6011c64618b : SUCCESS in 1m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/375d4c1ab5a7471e9079d3849ec70ef4 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d7a3df5f96543cc94808c609f2273bf : SUCCESS in 2m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d53ba756127f4d12b9c3baac8272a478 : SUCCESS in 2m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed166ce4f4fc4312a132f6b13fa98602 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc539a70372a439fba929f717c1fe328 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bafd008efb2e4b89abae2f91fecd0a3f : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55ed7bd08ffb4c0a893b3b1fa49252fe : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf20f2760a4c476baad429f717ae9e4a : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94498a8b4fb14be79e0db0e51056f826 : SUCCESS in 8m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcc79678ef1442949af7bc02a5a3b9c0 : SUCCESS in 4m 11s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/5fe92e1d51924c85ba3b7c3a94b558f5 : SUCCESS in 13m 37s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/ff25ab565df84150902e6f7baedce25f : SUCCESS in 14m 24s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0d4d53a27fd744fd83a83ba6f60a1d9d : SUCCESS in 43m 34s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0b989402adb041d49002491cc9bf2038 : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6126283ecf2c7d6b6d4c19df1b9df078cabca817","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 11:14:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6611f69f049d314390566fef6ae5446db54b2546","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 11:15:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c13906aa7d0479d9e1799611b58e69a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883a0bb49d9b4a8fb4399178642411d0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64cad81193aaf9bea9d4a930efc559288345ef24","revisions":{"64cad81193aaf9bea9d4a930efc559288345ef24":{"kind":"REWORK","_number":1,"created":"2025-01-07 00:04:53.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/25/938525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/25/938525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/25/938525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/25/938525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/25/938525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/25/938525/1"}}},"commit":{"parents":[{"commit":"10c1bb8c995efafb0395e2a2eba6245869e947e0","subject":"Merge \"Set up PBR with setuptools as a dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/10c1bb8c995efafb0395e2a2eba6245869e947e0"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-07 00:04:30.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-01-07 00:04:30.000000000","tz":-480},"subject":"Run python release package build job in check and gate","message":"Run python release package build job in check and gate\n\nThis ensures that we can package PBR properly before we try to make a\nrelease.\n\nChange-Id: Ibe1a8987d8f3752b9ab7b0991e00893e1cd840a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/64cad81193aaf9bea9d4a930efc559288345ef24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/64cad81193aaf9bea9d4a930efc559288345ef24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
