)]}'
{"id":"openstack%2Fpbr~938586","triplet_id":"openstack%2Fpbr~master~I6284e92b5214e7b816ae2a3a279f69160a029c0e","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-01-28 14:33:50.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-01-28 12:45:07.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-01-28 14:33:50.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-01-28 14:33:50.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-28 14:33:50.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-01-28 12:48:11.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I6284e92b5214e7b816ae2a3a279f69160a029c0e","subject":"Update setup.cfg example in Usage document","status":"MERGED","created":"2025-01-07 16:59:13.000000000","updated":"2025-01-28 14:34:56.000000000","submitted":"2025-01-28 14:33:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"938586","meta_rev_id":"0f654409c96a953308e04693b9c7cdfc11e97dcf","_number":938586,"virtual_id_number":938586,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-28 14:33:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-01-28 12:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-01-28 12:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-28 12:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-28 12:48:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 17:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 18:27:17.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-01-10 09:25:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-01-21 20:59:13.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-01-27 20:44:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-01-28 12:30:22.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"81b7114ba157eb8a04cd58cac7a5edbf6ec2bd02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-07 16:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f23eeba9387d34c2e20accdbaea4dd16667a587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 17:45:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5fd762a34fd4d83a364fb93a1c854d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7ab9e823b34417ae71d13781f9ac07 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc199f35cdfb4a178faec91888c38f42 : SUCCESS in 4m 34s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e8b1f2b8ef784a799b67108793eaac55 : SUCCESS in 24m 33s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/7fa19463f9154aeb8b2b4856e3ffaa5a : SUCCESS in 21m 39s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9aff40d7edcc4a8091cc969d5783d791 : SUCCESS in 36m 49s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/d2701f72b09e47008a6e738d85071ef7 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1a7bad06a2417d89241b58756c167fe8afcbc564","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-01-08 18:27:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b4d92c20641c88b5cd4c95af3ad2cb48c1dc8323","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-10 09:25:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c2cc1744d0f06d19145d5318af96494faac8ff1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-10 14:34:34.000000000","message":"Patch Set 1:\n\nYes, this change started out because I was noticing that it\u0027s a confusing example, so I was going to clean it up, but then realized the document did actually mean to use PBR\u0027s own setup.cfg file here verbatim.\n\nI agree that using PBR\u0027s config files as examples of configs for PBR-using projects is a poor choice, and if reviewers prefer I can work on a replacement example (or set of examples) instead.","accounts_in_message":[],"_revision_number":1},{"id":"5889a6538cd5abfb7bd29ec1886921622d404b44","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-16 16:26:05.000000000","message":"Set Work In Progress\n\nI\u0027ll rework this to switch to proper examples rather than an ouroboros of lazily reusing PBR itself as the example.","accounts_in_message":[],"_revision_number":1},{"id":"51faddcb1e0081527d3dd4a5eed8f977f359f0f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-18 22:20:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8aeab71278d3d44a750fc46f11a1512899341caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-18 22:47:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38fceaaa893c42faa13e64482bc4ea9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd6e04798fe4bf1904bbd845fe85dc0 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aefce72ed16482ab4a1e19de6cda5a9 : SUCCESS in 2m 17s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/bf76f86cb7474062963f5b1aa7d61786 : SUCCESS in 18m 29s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/7fa0f68a952045d0b1134c11b83ba7df : SUCCESS in 18m 57s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/d6c8e4f851f3413aa237d3bf6aa56863 : SUCCESS in 22m 27s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/e2bf88b3a22f4f7487efe96a96c66576 : SUCCESS in 24m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4efe9af398a056f32b80acc9d5153f23b6516190","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-21 20:59:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5a932795f0014ef805829d43138282720207a4fc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-27 20:44:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"615cdae3000329b3750e3058e70b35065e2dd8e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-27 22:30:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8173d8bf45929da364296f6b500e93a5d13c198c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-27 23:12:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2f3a2a7c83c4337a05f623472e40289\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031385698d3e4793b2d21ae1af868d33 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0464f2921d3499d9ce795ad828265fe : SUCCESS in 3m 43s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/58b956d821ea460595f58ceb88f83d28 : SUCCESS in 23m 44s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b0b007b1a02c4eec968ed72416b52891 : SUCCESS in 21m 23s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/d5a5709bf4e74953980c8e6f2699e119 : SUCCESS in 27m 48s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/68a1a0df62744c80a80296b4b3d25dc5 : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":3},{"id":"db5344ef38360fd8f7444afa3a79df1bf1abc6a9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 12:30:22.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"54c7885c0a1186cb7ef4d865ad4a4cf2eec08e69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 12:32:53.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"05153e65590eac604ccb4ea24b8d8e1455e0aa5a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 12:32:59.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"876f784d72da921dbfb3ed9de10cfcbdfd5244a6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-28 12:45:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ed40e73709f46d430ff9c0f6d7a9f6477ade352","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 12:48:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bf2b18f91e164a476937828b8deac67227972b46","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 12:48:18.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1cf8257e23609c507c10577be3a8982dc65d2eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 13:31:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d1fd22d6bc49c68e29e584bd0f96a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5310ef93d2b846d38aba064e5f6b6461 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20c5755016e74f37b21750cdefc060f6 : SUCCESS in 2m 47s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/8af11e0e96124306ac25184aa53a0ba1 : SUCCESS in 15m 42s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/46416aa280aa4efea95af32f8a2d564d : SUCCESS in 15m 45s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/1d5b08b8932c4d55acd18f83d348c5b5 : SUCCESS in 52m 14s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a81e8f192b774ac2970be6d425d7af44 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":4},{"id":"459f95c2ed6e98765c26791a189d2665326da5c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 13:33:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f9e00a6f0ea23eecca8e38b6e7d77ec1f7db013a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 14:33:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75f55725915040aeaecb5ce3022dc49e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e275c7dd6cbf426685bd7df99f6e3b23 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e425ad6e641e4c0a9f03a5d694e437fc : SUCCESS in 3m 22s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/69c8c9af023944fabf9f79e9fd8e06e8 : SUCCESS in 27m 44s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/db829308387a437689645a14bb0ad997 : SUCCESS in 24m 17s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/3315bdc40e714492bba355006707025e : SUCCESS in 39m 07s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/be95ea0325c64b1ea622052038839268 : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":4},{"id":"c382616ce582c4afc4dd8a9eb4aa3cc2903edd66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 14:33:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0f654409c96a953308e04693b9c7cdfc11e97dcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 14:34:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdb6019012ea49d38cd8e5cf64af0812\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545a0de21d3e41df859f8bf6a113cec3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6adf2f931300ae7f0ebb968ccb2d429d50670c07","revisions":{"198fc238b804de6600bebd1407a09a8236e01b45":{"kind":"REWORK","_number":1,"created":"2025-01-07 16:59:13.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/86/938586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/86/938586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/86/938586/1"}}},"commit":{"parents":[{"commit":"917a0196c4e9ed8f96fb5f911334e9fa47616411","subject":"Clarify PEP 517 implementation for users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/917a0196c4e9ed8f96fb5f911334e9fa47616411"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-07 16:58:14.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-07 16:58:14.000000000","tz":0},"subject":"Update setup.cfg example in Usage document","message":"Update setup.cfg example in Usage document\n\nThis replaces the prior example with the verbatim content of PBR\u0027s\ncurrent setup.cfg file, as the two had diverged long ago.\n\nChange-Id: I6284e92b5214e7b816ae2a3a279f69160a029c0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/198fc238b804de6600bebd1407a09a8236e01b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/198fc238b804de6600bebd1407a09a8236e01b45"}]},"branch":"refs/heads/master"},"197ee141621df34b6ff9176362acc2faeba33475":{"kind":"REWORK","_number":2,"created":"2025-01-18 22:20:32.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/86/938586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/86/938586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/86/938586/2"}}},"commit":{"parents":[{"commit":"917a0196c4e9ed8f96fb5f911334e9fa47616411","subject":"Clarify PEP 517 implementation for users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/917a0196c4e9ed8f96fb5f911334e9fa47616411"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-07 16:58:14.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-18 22:18:59.000000000","tz":0},"subject":"Update setup.cfg example in Usage document","message":"Update setup.cfg example in Usage document\n\nThis replaces the prior example (an outdated verbatim content of\nPBR\u0027s setup.cfg file) with a clearer fictional one.\n\nChange-Id: I6284e92b5214e7b816ae2a3a279f69160a029c0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/197ee141621df34b6ff9176362acc2faeba33475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/197ee141621df34b6ff9176362acc2faeba33475"}]},"branch":"refs/heads/master"},"895f6fb3812cdad88a177a256928cd0a8caa0664":{"kind":"REWORK","_number":3,"created":"2025-01-27 22:30:38.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/86/938586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/86/938586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/86/938586/3"}}},"commit":{"parents":[{"commit":"917a0196c4e9ed8f96fb5f911334e9fa47616411","subject":"Clarify PEP 517 implementation for users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/917a0196c4e9ed8f96fb5f911334e9fa47616411"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-07 16:58:14.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-27 22:30:33.000000000","tz":0},"subject":"Update setup.cfg example in Usage document","message":"Update setup.cfg example in Usage document\n\nThis replaces the prior example (an outdated verbatim content of\nPBR\u0027s setup.cfg file) with a clearer fictional one.\n\nChange-Id: I6284e92b5214e7b816ae2a3a279f69160a029c0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/895f6fb3812cdad88a177a256928cd0a8caa0664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/895f6fb3812cdad88a177a256928cd0a8caa0664"}]},"branch":"refs/heads/master"},"6adf2f931300ae7f0ebb968ccb2d429d50670c07":{"kind":"REWORK","_number":4,"created":"2025-01-28 12:32:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/938586/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/86/938586/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/86/938586/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/86/938586/4"}}},"commit":{"parents":[{"commit":"917a0196c4e9ed8f96fb5f911334e9fa47616411","subject":"Clarify PEP 517 implementation for users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/917a0196c4e9ed8f96fb5f911334e9fa47616411"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-07 16:58:14.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-28 12:32:53.000000000","tz":0},"subject":"Update setup.cfg example in Usage document","message":"Update setup.cfg example in Usage document\n\nThis replaces the prior example (an outdated verbatim content of\nPBR\u0027s setup.cfg file) with a clearer fictional one.\n\nChange-Id: I6284e92b5214e7b816ae2a3a279f69160a029c0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6adf2f931300ae7f0ebb968ccb2d429d50670c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6adf2f931300ae7f0ebb968ccb2d429d50670c07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
