)]}'
{"id":"openstack%2Fpbr~940117","triplet_id":"openstack%2Fpbr~master~I2f7d1a50237ff9f63ad7cfd145078d850921a228","project":"openstack/pbr","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-01 11:07:50.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-02 20:27:26.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-02 15:02:54.000000000","reason":"\u003cGERRIT_ACCOUNT_5263\u003e replied on the change","reason_account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}},"hashtags":[],"change_id":"I2f7d1a50237ff9f63ad7cfd145078d850921a228","subject":"Add openstack-tox-py313 job","status":"MERGED","created":"2025-01-24 10:30:05.000000000","updated":"2025-07-02 20:28:53.000000000","submitted":"2025-07-02 20:27:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940117-py313","meta_rev_id":"b61675a5f9140a46332824f42e23431996b213ba","_number":940117,"virtual_id_number":940117,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 20:27:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-07-02 17:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-02 15:02:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-02 18:37:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 12:16:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-24 14:27:02.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-01-24 18:13:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-05-04 07:54:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-07-02 17:27:21.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"5fec960bf336ca85783e99de2bc3312fc72710ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-24 10:30:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4858baa4a213d5723a324c0043f8047aff4cdf86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:16:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf37367e096b40e9b3e453029c164267\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/faf5870eee2746489b2f308686ef5c6d : SUCCESS in 1h 34m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0af8e4370c4190a43b3f1d0827abcc : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7392d1494e44bfd8ad707b6c3ab2ea8 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67c4dda73cb14ee6b3e3ff1c538267a3 : SUCCESS in 6m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32ff1d16cbb345ee99d043fee7c5994f : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31f821ae543f42a49134279ba3a9f42a : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a2e52f8114b4833afded09cef588dcc : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ebf5657495b4be68a7b6119539ab203 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68a08070d7314fe199d1d07eae33a7be : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a146edbb25c9440ebe3c8aa119b7bf64 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e991d02b02e2429cb1cb5c188296552f : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11ef0e98409f4ffcbda9978fef2abde0 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d57f006e1b548f3954d9780c7adf346 : FAILURE in 5m 30s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/3b08e491191647d59dff24d15deecaf9 : SUCCESS in 48m 17s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/3f6b8613d0fa4bebbeb94b1b4af4f41a : SUCCESS in 34m 09s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/925d71ad33654141817bc507ddc0ebdf : SUCCESS in 34m 21s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/f503b7cbdac64533b554a2eae7fa57c3 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fca5c57f01361703dd733f194faba77f80978741","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-24 14:27:02.000000000","message":"Patch Set 1:\n\nThe failure is vague, but I think it\u0027s saying that py313 doesn\u0027t match the 3.13.1t installed in the job? Looks like our pyenv role might be naively matching on t-suffixed versions now that those exist in upstream CPython:\n\nhttps://zuul.opendev.org/t/openstack/build/2d57f006e1b548f3954d9780c7adf346/console#3/0/11/ubuntu-noble","accounts_in_message":[],"_revision_number":1},{"id":"3849699024b88dd5067b10b402b2700faec93d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-24 17:07:42.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d2ab2a08ce4eb3fc0c616acb03f282e725832bb0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-24 18:13:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"02fcf1498c4245f287ed4cfe81c1a9b3055e919f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 19:36:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da302369c5864c109f2d169f2d3191ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8172bd299dc4e4695a67cb98ba2219e : SUCCESS in 2h 21m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21e921b36f2401f8c424801667194bd : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca58d5550ab4fecadc908d1a1f97a14 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc1c00ed3d814597bf89ac03995ab9a9 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a1ad2207cfa4c50b98d125a1acc4af6 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b372bef330934057aff39813630e9474 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9aaec2848802453abebe1f87252d8259 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9ecd250e778463a8ac591efff8dcaed : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af5636b51e5c49919eacb3ad5ec83d38 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45e9c7ea005f4da79548bfea1b560901 : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5b42d1f77974ee68b6a7a8753860d2d : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/659c8501d91b4ba58b0b88cd9a7e1a4c : SUCCESS in 6m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e9cd0b6d329482aa71d65616442dba1 : SUCCESS in 10m 30s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/118b7f3b53214432b29e6468f822a11a : SUCCESS in 21m 11s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/005889f6298a464b8aa262f1b6be2608 : SUCCESS in 55m 23s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/8f0b23de870f4d76a399dec3d1eb0a5c : SUCCESS in 30m 51s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/100c3a2e365c4df783a2c1fdaddfaf98 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":2},{"id":"15924ea7c3b631158a8109f5e69fea6bf9cb75b1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-04 07:54:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8dcc59069dda66637fc9ea3ab9901fdb89838a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 08:19:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7fd3e3fde404bbe9fd9924798b9eda4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/197ee7105e424cd5ad6e24bdd39d7c1a : FAILURE in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d931709bfb3440baf46586a94b3bd65 : FAILURE in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d73062fdd04c1b8ec2970e0e4a3f81 : FAILURE in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a01c5e63262480cb3680b11c041e2d7 : FAILURE in 1m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/695092354c3a47a0b54dbad014926972 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d66b73c3fa4d48a8aec04f9cb8d9446b : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd6f24eeca6e4d3fb9429c072f81532e : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/425f4696307243e38b31e42a7fce2f15 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e13e4fdef21642f987aa101dae661fa8 : FAILURE in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5750be1cef4e4c6d8902b645f5e0ddea : FAILURE in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/969890f5eee549e68669b5540d2fa599 : FAILURE in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cadb6b5fe00447a7adafce4d9173f396 : FAILURE in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df65855eb3854169b991e0f51f23f199 : FAILURE in 4m 18s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/f3372e8f420d4d899ea49d20ecfa86c4 : FAILURE in 11m 16s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6fc9a103238d43128e7c1b90b41162b1 : FAILURE in 10m 53s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9b9bcf2e52de4d98a9baa5e2c15757de : FAILURE in 15m 20s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/2dad6df2185247a48edc1ceae4de4633 : FAILURE in 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"b691cb2e0bff2c0c6bb37fb4a239146f325bbe2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-04 08:40:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"260e755eaba2f44a897ede38ec2705bedb023e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 09:08:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21a71bc9c73c49c3ba35b22ee36587a9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31dfa8a8a32e42a5ba5e9191bf0403ae : FAILURE in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70abfc67abb6489c8f1d485117e95168 : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f40ba8297fd4f778c3d5af7366fb121 : FAILURE in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27818cc6ec864a718611fa0284b8cb71 : FAILURE in 2m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/897f86f64c8b4461b2544f6b6801547d : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4361d333733d4a1398e96680fceb7551 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b7979584d14420aa86761ef02192ec8 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9f303738dda4efd8ecf13a21a5f07b3 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52b96f7e7206427f9f9480c8bfced7b2 : FAILURE in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b1842baa4c040c0b734d21f193eb49d : FAILURE in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b69267f2041426b8c920d35b45ba18d : FAILURE in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d31f27b819ac473ab2c16b799e61bb09 : FAILURE in 2m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/247b77da3f1144ec98d9defd6f66812e : FAILURE in 4m 19s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/fc86881edc7747c9b7de1de86bb43699 : FAILURE in 11m 12s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/ce9900adf9114d3dbf487e4ccc7768c5 : FAILURE in 10m 43s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/a8d224818bb34acf8b7b19b1f838c940 : FAILURE in 15m 34s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/bfe721e6d50d4721b8158eb23cfced8a : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4556e4cb2195f1cd9746e4da07b82d7bc7763be1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-05 08:17:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60157eaf6af2a7206925c06d391bec282f9c5a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 10:13:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2769a5ee5f834b14aa55eb3451c026de\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f84d4e2ae4ad4f948e710c84bab63d21 : SUCCESS in 1h 35m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f4cb4c0540486bb392e3f41cfc6635 : FAILURE in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae0b528a6eca4e1fa10b7a28700b4434 : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7a291bdfb644106aac3daba09c2385a : FAILURE in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3a6872575844f3196c81ca001a1e811 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c539985e6a74a5eb73fe3ac3265f06e : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c826cff2283a4fb7a53e55db11179e21 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d6e568d125e400faabe9b9e52fc7741 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0bd973652b49cc85fc398d6b76d832 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66c4d7fcc05b4a459ab14ca7be8a5550 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6327dbdf3377400c86d434abde430525 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93c1207a421e4344b031e9ebfa9f2bf4 : FAILURE in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/439b2525d27545678be5e93cbda2c392 : FAILURE in 7m 29s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/352c1b15e470415788aa4ad2c6597da3 : FAILURE in 27m 49s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/a6c0bfc3aab844dcb8298ba265a3d359 : FAILURE in 26m 35s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/85d2831eb2ea4dcdb1e9b4d4c9eb1c09 : FAILURE in 13m 42s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/d91fdb48bd354c658f025eedaa0c5cd9 : FAILURE in 15m 29s","accounts_in_message":[],"_revision_number":3},{"id":"388753b253822c023473001d4c2aba407acd081d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-06 05:27:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"355d9ecbfac36d757a0ec29c689ffa877e1b316e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 07:54:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d264a73f4ebdb0f2905e36309ae70da13f9609f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:33:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e17e503747344ad9d5212889dcf1798\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6142f3ce9aaa40d5a7afeb22e66cab3f : SUCCESS in 1h 33m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95238d57d2a463d850906260cedc23f : SUCCESS in 1m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f39c158bb1314616b8710a9f0db117d8 : SUCCESS in 1m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ad2f5cd7789413ba8a4d181556218c3 : FAILURE in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2fde0478126f46f4abaf5843dc3265ee : SUCCESS in 2m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8806b2e282a4b90ab74c398db44192d : SUCCESS in 2m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7dff453d33e4824acede42f0065fbe5 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82bb81365dfb455aa62d91c8ded91268 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15f2b2efcd7641ffa7debc8ab42574d8 : FAILURE in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cb3c61598ec4692ad19318315860885 : FAILURE in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6adfb284c9d04dc1be47bfa78659fc7b : FAILURE in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21a69d22eb5e4c4cb17ee52e04c33ed2 : FAILURE in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4005d8c9e3154744ad8c7bce581ffa28 : FAILURE in 7m 19s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/eca178590705429285b332d42d4c7488 : FAILURE in 13m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/155955c8b29940f0bd690139372ec447 : FAILURE in 12m 54s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/f074bc1aced841fc83b86761ffa9d54a : FAILURE in 6m 52s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/7f02fb9cb8ed498ab17d97ff543b00c8 : FAILURE in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"ee0aed5a42f03470d0345db495f4bac0c0f05ba1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-30 18:49:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"253f1aedb5b38c3bc8bf919e6fc76f289e46e68d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 20:29:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b6b290faf6743d69ca0d2c873af2ebc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1db7cde1cf944058785d4990b1caa71 : SUCCESS in 55m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e059937499b04c26b3a8dbd8628686ed : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4efbf41f47942c1ab39658c93c1e9ee : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa4330a43c9431c85d46790c4fbf9e3 : FAILURE in 8m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fcd6e096e7e4ace8ba6248647140a7d : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b2019b46aef421d85c17bb2e0f04a8c : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cee51f5cf9504e48991050517734bdae : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb5df63c76f64b4b93069071550c2f09 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff854e56670c4acbb2eaaaf250d55e77 : FAILURE in 7m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15169b13849742daa7e9f6fff5751c1d : FAILURE in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/443e7598c0024cf7aa032990bc6783d4 : FAILURE in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ce65f5af8494e8b99d26428e6f08936 : FAILURE in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75d99892afa14645956d6447621b8bd5 : FAILURE in 5m 24s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/6be77291ac6343f0a379f2307d78a962 : FAILURE in 24m 02s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/207a80ab9d194dbea9c787eb18de2164 : FAILURE in 32m 45s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9b997c08684043939bdd00977942a485 : FAILURE in 20m 36s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0cdaee4d1a804d9881e09b8d8d99a36e : FAILURE in 8m 41s","accounts_in_message":[],"_revision_number":3},{"id":"829ec3ca8c32b04c5e46142eb27bb43b6915de0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-30 21:24:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"96b4ec1f6c3b4df13aa29a503016f27a84fe89c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 23:17:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bd92798c73e46c89ce176bc64ef6897\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/694f032fe1b1467f930309f3271e4423 : SUCCESS in 1h 38m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f9ef8dea8f4dc6886ece7b9116aa44 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dedadfd96f93486580594b5dc2ea803e : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e3a7f2441574d22b2aa96eeb41108d7 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f9607d2d3f343599ce50456fa385662 : SUCCESS in 2m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d11a4d7768924c858eb52658601c3ab3 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/620b14d36e6f43708f5484d4455e897b : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0df212d1d4ac4af6a2d678b739552b48 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d807fc41e02421e904d9de20f417b70 : SUCCESS in 2m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96eefe15e64b4b9d982c1b6c63c3d115 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00ec8e93a0df413996f52858ac9a7b52 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f43d4cae48d746bd86fcb1ef291cf7e8 : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/edac126340484f6284e5e168f726f4f8 : SUCCESS in 8m 50s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ff48fa224fbc4fe5aa2363bbe6b3ae70 : FAILURE in 24m 43s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/421413a548574f00ba99d2400774da7f : FAILURE in 25m 09s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/872f67c70802411a95654614c88a80ba : FAILURE in 12m 29s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/121e391ccb494cb881d82126c70af7cf : FAILURE in 13m 49s","accounts_in_message":[],"_revision_number":4},{"id":"3766958d97ea5199f5170ef6ba5c7c95e37c7287","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 12:52:28.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8ceaabc27220450d6c9cf893b9a06b2991945ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 14:15:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/002185788cb54c91ae8661267ea72c59\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbaf021a9f2d4557958c73e0613d5cbd : SUCCESS in 1h 20m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f1b386f5b44960962d24c0750729b4 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc76cd8795c0431a93c657021370bfae : FAILURE in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4db7962f2d2447629945252b8bdf8850 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1f623b32de64c098f6dc9ef7d557f72 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e33ee511c94b7d897fa5574d275e00 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da0407d443a945cb84db57f7bb1970c1 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b32b73f4ee42485db89f6913fe485937 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e7863cb93e44cf69d6b23d27c189666 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25ec381c56f247d384dbf203ab89c2c3 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6920ccf30f0f41eba6e7a20e00ced7ae : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fdbf132b14e4c15913b3b3128e91a6b : SUCCESS in 7m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0859c47fdf24348a3adc80f19de678b : SUCCESS in 9m 21s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/aa9fc10c9dfc43bd87c3e2dc97c9a104 : FAILURE in 29m 46s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6753066dbee94445a75827e19a3bdf5f : FAILURE in 25m 43s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/b608a1c36d0043bc958ba0614566de4a : FAILURE in 7m 01s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/35604a23f4cd4fd4bffb93db4ca5ae8b : FAILURE in 15m 22s\n\nWarning:\n  Comments left for invalid file pbr/tests/test_integration.py","accounts_in_message":[],"_revision_number":5},{"id":"73e8b76038ef2d06bca18ee9edb7d2274edaf7af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 15:58:19.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"746a67cbd5ba8ec7fd3c3a10bd2bf6db9c004e25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 17:24:01.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"4b886b7db65e93a7cf58bf3cb5a2dcbc26390ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 18:28:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3aa1f36e7457424895125ceef21265c3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cabb2e31c12c4e478981d2cbfb4e1585 : SUCCESS in 1h 00m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaede194ef1e4cacbdcfd7849ce056c0 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5949466a789b46208badcfa46ae00538 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/073666b8d63d4b15bf757428c8c72871 : SUCCESS in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5a00fb5bc244d019a1237d4a9966906 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecfe8e836be8490295e82bb89ff5f57f : SUCCESS in 2m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6f38cd2453b421fbc0eb82fb750b9dd : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6877425e6a0f4b93a6f9efa8830fd9a3 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7e6d0663bfe449381f24603958be7be : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12d0d0386a01480bae5f31701bdaf764 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a425abd355e4d48bdee5d3d7037ccc4 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3397ebd1bab4babbf882a8cc15b9986 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/959dbd284dbb4516987c743327527be1 : SUCCESS in 11m 01s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/cd0499208e4f40198abcd0e66211955e : SUCCESS in 17m 15s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/5c7e01011f074e89964cf2c6bb06bccb : SUCCESS in 27m 33s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c15c5db8ccec424d8f515b86a50f1fe2 : FAILURE in 8m 48s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4ea54ffba7974dea932f47a83cb2e091 : FAILURE in 7m 04s","accounts_in_message":[],"_revision_number":7},{"id":"afa387fd7fceca803449fe566c8586f6bf653fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 21:58:52.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"7f1ed94497f2ee35e9e5a1d781b59b183430452d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 23:46:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc9efa2c78234a08b9a72d50442591ef\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7865aa232127469ea93fdab5fa8b49c5 : SUCCESS in 1h 36m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55e88bad35b40bba10163eff58aa652 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ff748bf7db4ce7a1ad732a10f72d4f : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2537d3e1da814e039820b3607c25cab5 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a10809214b4c474bb11252cd4c38efca : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d659bb52554900b8015049ae9fefb8 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcdb637195ef4665a221ee5d2cc86c18 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b854370e16f40a3a54a59edd92e7216 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/245590c4d4004944bd3b9333ffab0e3f : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7553e146d6d34021a869a377423ed19b : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/902080915eac46c891650ac91493e22e : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f919b4185834ad8b3f48e9a22254c0c : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1c7a60537d9484da369f5febaf47189 : SUCCESS in 11m 56s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/2d1ec942824b40f59193d501032f79b2 : SUCCESS in 24m 55s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/a2a72f162823444e8287873512a1e96b : SUCCESS in 27m 33s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/d5635b06f3d34a03938419dba70bb0d2 : SUCCESS in 29m 27s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/5da7eb4602434f74a856278838bb45fc : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":8},{"id":"cecaaf9dc2c24d52b96478a7832c8152d91ab229","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-02 15:02:54.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"2fcb6b23aff8ab43ecfe926eec51898977a2a578","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-02 17:27:21.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"10796198cd490c9cb8919374540e14a4d376481c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-02 18:37:00.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"7ade79583ef04256ebc10b91935a831d3330bb4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:38:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5b2789ab7e8dd221eb1004845bca133c47c054ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:27:26.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaa005f73b7f4118b7dc38f0cb4ee394\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc8bbeb586414216927eba001b48d0af : SUCCESS in 1h 37m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0671adf1f5f945ccbc254ddf3b7176fb : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9a37a5f71c457185250d69c5e22ca2 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72c8394c94204376953c08442fcc3931 : SUCCESS in 9m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd29771e755543b09733d3466073579b : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7590c144246241d5ab1c5bbc4130516a : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/540baa0083aa44f28a58c5ff0e75177f : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/094788a9581f42ee92de159a1550f5ae : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8259f0c9dc4ac1a56bd0aa7b91097a : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ebe982482254edd855897b73abe3fa2 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20d7d3bc18604cddb052588190378711 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b566fed609247668a10cbb995f3bfc9 : SUCCESS in 7m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/590a5ba831e14ec98e197b78c6939a9a : SUCCESS in 11m 43s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/8f8596ba79214052b493489141e8404e : SUCCESS in 26m 36s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/da1ab9784d00414582557aeea2482998 : SUCCESS in 25m 22s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6b5552827b664fff9b8d6436059ab9e6 : SUCCESS in 34m 19s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/5927bc08dd914f8882091888c55a6457 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":8},{"id":"535812bee41a110b46bf5e24d0aed56e7a8c7df5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:27:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"b61675a5f9140a46332824f42e23431996b213ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:28:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb04bbdc41c48ee80c9040499d2d54d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1857fa22accd4dccb34dfa8ca2ea9884 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"141d0eac9e10aea033f1f6fe4d255ddac28ca3c4","revisions":{"89577931507146b8b243d5ba645903605838e116":{"kind":"REWORK","_number":1,"created":"2025-01-24 10:30:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/940117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/1"}}},"commit":{"parents":[{"commit":"917a0196c4e9ed8f96fb5f911334e9fa47616411","subject":"Clarify PEP 517 implementation for users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/917a0196c4e9ed8f96fb5f911334e9fa47616411"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:42.000000000","tz":0},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/89577931507146b8b243d5ba645903605838e116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/89577931507146b8b243d5ba645903605838e116"}]},"branch":"refs/heads/master"},"0d9211d5433ad7b5e0e50d7b33cb52b02730b704":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-24 17:07:42.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/17/940117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/2"}}},"commit":{"parents":[{"commit":"917a0196c4e9ed8f96fb5f911334e9fa47616411","subject":"Clarify PEP 517 implementation for users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/917a0196c4e9ed8f96fb5f911334e9fa47616411"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-24 17:07:16.000000000","tz":0},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/940158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0d9211d5433ad7b5e0e50d7b33cb52b02730b704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0d9211d5433ad7b5e0e50d7b33cb52b02730b704"}]},"branch":"refs/heads/master"},"a4ee66bf8c68a2cd394e590c6bc155068d208d25":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-04 08:40:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/17/940117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/3"}}},"commit":{"parents":[{"commit":"5d4a1815afa920cf20e889be20617105446f7ce2","subject":"docs: Drop SetupTools from pyproject.toml example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d4a1815afa920cf20e889be20617105446f7ce2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-04 08:39:29.000000000","tz":120},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/940158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a4ee66bf8c68a2cd394e590c6bc155068d208d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a4ee66bf8c68a2cd394e590c6bc155068d208d25"}]},"branch":"refs/heads/master"},"911c38772dbe14ff495355b08022254aab902077":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2025-06-30 21:24:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/940117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/4"}}},"commit":{"parents":[{"commit":"838484101902af66633d4086c570e6db11fc3d7b","subject":"tests: Set GNUPGHOME in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/838484101902af66633d4086c570e6db11fc3d7b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-06-30 21:24:19.000000000","tz":60},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/911c38772dbe14ff495355b08022254aab902077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/911c38772dbe14ff495355b08022254aab902077"}]},"branch":"refs/heads/master"},"599efa5cbc0b315a83f808d395558cd17bdd78d4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-01 12:52:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/940117/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/5"}}},"commit":{"parents":[{"commit":"447cd9116d697abaff32c6c4489f6bb4cc1e9426","subject":"tests: Set GNUPGHOME in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/447cd9116d697abaff32c6c4489f6bb4cc1e9426"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 12:52:23.000000000","tz":60},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/599efa5cbc0b315a83f808d395558cd17bdd78d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/599efa5cbc0b315a83f808d395558cd17bdd78d4"}]},"branch":"refs/heads/master"},"fd5d5866a3e9eac5e62e684b77871298ff1dadfa":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-07-01 15:58:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/940117/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/6"}}},"commit":{"parents":[{"commit":"a7cf353d1f7d41436bd73aa5da72ec5322ac2458","subject":"tests: Set GNUPGHOME in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a7cf353d1f7d41436bd73aa5da72ec5322ac2458"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 15:55:59.000000000","tz":60},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd5d5866a3e9eac5e62e684b77871298ff1dadfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd5d5866a3e9eac5e62e684b77871298ff1dadfa"}]},"branch":"refs/heads/master"},"496bacc7f654b1e4102805df6ff3e98649d277df":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-01 17:24:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/940117/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/7"}}},"commit":{"parents":[{"commit":"d7f01d78b15aee17bfde96d9598186d03860883a","subject":"tests: Set GNUPGHOME in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d7f01d78b15aee17bfde96d9598186d03860883a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 17:23:55.000000000","tz":60},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/496bacc7f654b1e4102805df6ff3e98649d277df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/496bacc7f654b1e4102805df6ff3e98649d277df"}]},"branch":"refs/heads/master"},"141d0eac9e10aea033f1f6fe4d255ddac28ca3c4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-07-01 21:58:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/940117/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/17/940117/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/17/940117/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/17/940117/8"}}},"commit":{"parents":[{"commit":"90de90c27642cb8976f83a34f15b7b547ea3eeb3","subject":"tests: Set GNUPGHOME in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/90de90c27642cb8976f83a34f15b7b547ea3eeb3"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:29:25.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 21:58:47.000000000","tz":60},"subject":"Add openstack-tox-py313 job","message":"Add openstack-tox-py313 job\n\nChange-Id: I2f7d1a50237ff9f63ad7cfd145078d850921a228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/141d0eac9e10aea033f1f6fe4d255ddac28ca3c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/141d0eac9e10aea033f1f6fe4d255ddac28ca3c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
