)]}'
{"id":"openstack%2Fpbr~940311","triplet_id":"openstack%2Fpbr~master~I015b271335b9419ac5a73f7f19320087dd5eb1a2","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-01-28 17:37:41.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-01-28 20:09:39.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-01-28 20:09:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I015b271335b9419ac5a73f7f19320087dd5eb1a2","subject":"docs: Drop SetupTools from pyproject.toml example","status":"MERGED","created":"2025-01-28 15:13:36.000000000","updated":"2025-01-28 20:11:14.000000000","submitted":"2025-01-28 20:09:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940311","meta_rev_id":"9944e8c755473056a0b9acb34d58cb401fa3a76b","_number":940311,"virtual_id_number":940311,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-28 20:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-01-28 17:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-01-28 17:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-01-28 17:37:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-28 15:21:43.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-01-28 16:32:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-28 17:36:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b46c136e8597ca07ac77af0bfe9aa71e53a91b23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 15:13:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9ff9b60b6854198741089950195e6507168ace3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-28 15:21:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ba5e1d2c16dce620d3826e8d1a130a9017f06880","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 15:25:36.000000000","message":"Set Work In Progress\n\nThis may need revising to reflect the release of PBR which includes the dependency declaration on SetupTools, in order to work correctly on Python 3.12 and newer.","accounts_in_message":[],"_revision_number":1},{"id":"e2676a8e9d3ac69a49fec8c8332c8f5995166f95","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 15:30:33.000000000","message":"Patch Set 1:\n\nRevision 6 of I2f7940e14a1d18dec13a15e3ef66af0381e515f5 demonstrates this as working with the current 6.1.1.0b1 pre-release of PBR.","accounts_in_message":[],"_revision_number":1},{"id":"be1f32409c16852f45b1b6d8f755271b3e6f39c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 16:32:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c57a87ddd3c4f508f50c52a4c9fca55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c96f95fc344cbeaf59315b7e215b48 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99825cdc4b944e309a2e40f61d10edde : FAILURE in 3m 46s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/6ed8e905519d40108b47fce6cd3de4f6 : SUCCESS in 55m 28s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/13901b47cfd24d5697d82ab803f42992 : SUCCESS in 25m 19s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/74ac7a2f61b54d2ca687fde47f8030b8 : SUCCESS in 37m 28s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a2c393577ca240adb2c47b873e4e182d : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"72bb7936f52a915daa6fe68d190eeacdd9393303","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 16:50:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7dc08a14e5bd6c1d865f7ebfef6e0d58312c308a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 16:51:45.000000000","message":"Patch Set 1:\n\nAnd from the failures on I2f7940e14a1d18dec13a15e3ef66af0381e515f5 it seems we\u0027re going to actually need the parent change in a release first after all, because the ephemeral isolated build venvs created by pyproject-build aren\u0027t including setuptools by default on older platforms.","accounts_in_message":[],"_revision_number":1},{"id":"3e2e1ce99961e430110d1a30a301db4c4cc0f80c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 17:18:01.000000000","message":"Patch Set 2:\n\nDigging into the https://github.com/pypa/build source, I\u0027ve confirmed my observations. In PEP-517 cases it intentionally omits (and if necessary removes) SetupTools from isolated build envs regardless of Python version, while in non-PEP-517 cases it installs SetupTools always for backwards compatibility.","accounts_in_message":[],"_revision_number":2},{"id":"b7301eda1364af8ffaaed8f761fdb2d6ea51e570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 17:35:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad215531dc9e4bf3bb1919e8854f5040\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387baa39ef644f70aad11261111c54c9 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64beef87ae034f9b9210a67b945dbc00 : SUCCESS in 3m 36s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/9bc9884f521d46b7944c97b442b37696 : SUCCESS in 26m 02s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/3bbdc8b8e1ff4406bfe63f01eed47cd3 : SUCCESS in 22m 47s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0f59e95675cc4130b2246720ad30fcf1 : SUCCESS in 37m 17s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/9169f1c722da4ff3a995b7da7416bd9e : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":2},{"id":"80a7107cd32277b32da71ce7db27bf93a78d426f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 17:35:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0cb52c5bef3910d70ad985d498e71b43c0295708","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 17:36:46.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0f293dd769b32a7ceb5520fd6659b1aa66a3b3b4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 17:36:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9b89cfc8c049c6ed27f1fdd48828897025639e4f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-01-28 17:37:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"87b68cd5a605aaf5e4c899e27815efc0b33580ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 18:26:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d213dbcb27e4e0081ae1ced0560d147\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50cdfaf9713047a2957fcbbdf4bfe317 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d13d6ae1e5f1420e9473e6de4d5e4f78 : SUCCESS in 4m 47s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/f5063227fa5843bb85a39b45a40e36db : SUCCESS in 24m 08s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/c318b3959d644d4cbf9f56ed599c2e03 : SUCCESS in 16m 10s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4942dccde74043498c0ffe66b089eca5 : SUCCESS in 35m 13s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/57338df6bf8848f49e72b08dcad5fbc0 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":4},{"id":"286292d86bf3a4caf4cf72bdae79054ab0c03d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 18:31:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"60d56792f194585950407a5fe1db0e071239d36d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 20:09:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/854b1e29c8804a7c815927f908c584b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1726d63d5440498d1619272617e196 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d40f261ea1496792970e1e82e6428e : SUCCESS in 3m 52s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/f0805c6a26724d38bee1adaffe1b415f : SUCCESS in 29m 16s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2a819eb405284935858f0bb0a936822e : SUCCESS in 23m 09s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/d1b9e6ee44bc42468234b5d539c93533 : SUCCESS in 27m 41s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/95ffc25f46784aebae5466e3a965aa1a : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":4},{"id":"c40d236a6b626970475780ac0d164e70a1663023","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 20:09:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9944e8c755473056a0b9acb34d58cb401fa3a76b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 20:11:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/196fdc8956a34bd180f0a16e6ea609dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7307c3e16d942aca78a12ac3d6f6822 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5d4a1815afa920cf20e889be20617105446f7ce2","revisions":{"07cc0f93051ea37ba40fc23ed88a90950cff59e9":{"kind":"REWORK","_number":1,"created":"2025-01-28 15:13:36.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/11/940311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/11/940311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/11/940311/1"}}},"commit":{"parents":[{"commit":"1e73fb4827dda1006d8b5b81772b2fefd8cd91e9","subject":"Simplify PBR dependency on SetupTools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e73fb4827dda1006d8b5b81772b2fefd8cd91e9"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 15:10:11.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 15:10:11.000000000","tz":0},"subject":"docs: Drop SetupTools from pyproject.toml example","message":"docs: Drop SetupTools from pyproject.toml example\n\nSince PBR 6.2.0 and later declares its own explicit dependency on\nSetupTools now, there\u0027s no need to specify it separate from PBR in\nbuild backend settings. It should \"just work\" everywhere.\n\nChange-Id: I015b271335b9419ac5a73f7f19320087dd5eb1a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/07cc0f93051ea37ba40fc23ed88a90950cff59e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/07cc0f93051ea37ba40fc23ed88a90950cff59e9"}]},"branch":"refs/heads/master"},"1563fa99e2e90f847c16b09c524672f6486d576b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-28 16:50:20.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/11/940311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/11/940311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/11/940311/2"}}},"commit":{"parents":[{"commit":"28eee6c16b0e84d6264b97eac8fbdff49dc2bc03","subject":"Simplify PBR dependency on SetupTools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/28eee6c16b0e84d6264b97eac8fbdff49dc2bc03"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 15:10:11.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 16:44:51.000000000","tz":0},"subject":"docs: Drop SetupTools from pyproject.toml example","message":"docs: Drop SetupTools from pyproject.toml example\n\nSince PBR 6.2.0 and later declares its own explicit dependency on\nSetupTools now, there\u0027s no need to specify it separate from PBR in\nbuild backend settings. It should \"just work\" everywhere.\n\nChange-Id: I015b271335b9419ac5a73f7f19320087dd5eb1a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1563fa99e2e90f847c16b09c524672f6486d576b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1563fa99e2e90f847c16b09c524672f6486d576b"}]},"branch":"refs/heads/master"},"78786bd8f09854315c543ea7241fd907ab20e01c":{"kind":"REWORK","_number":3,"created":"2025-01-28 17:35:52.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/11/940311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/11/940311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/11/940311/3"}}},"commit":{"parents":[{"commit":"28eee6c16b0e84d6264b97eac8fbdff49dc2bc03","subject":"Simplify PBR dependency on SetupTools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/28eee6c16b0e84d6264b97eac8fbdff49dc2bc03"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 15:10:11.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 17:35:34.000000000","tz":0},"subject":"docs: Drop SetupTools from pyproject.toml example","message":"docs: Drop SetupTools from pyproject.toml example\n\nSince PBR 6.2.0 and later declares its own explicit dependency on\nSetupTools now, there\u0027s no need to specify it separate from PBR in\nbuild backend settings. It should \"just work\" everywhere.\n\nChange-Id: I015b271335b9419ac5a73f7f19320087dd5eb1a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/78786bd8f09854315c543ea7241fd907ab20e01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/78786bd8f09854315c543ea7241fd907ab20e01c"}]},"branch":"refs/heads/master"},"5d4a1815afa920cf20e889be20617105446f7ce2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-01-28 17:36:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/940311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/11/940311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/11/940311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/11/940311/4"}}},"commit":{"parents":[{"commit":"28eee6c16b0e84d6264b97eac8fbdff49dc2bc03","subject":"Simplify PBR dependency on SetupTools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/28eee6c16b0e84d6264b97eac8fbdff49dc2bc03"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 15:10:11.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-28 17:36:46.000000000","tz":0},"subject":"docs: Drop SetupTools from pyproject.toml example","message":"docs: Drop SetupTools from pyproject.toml example\n\nSince PBR now declares its own explicit dependency on setuptools, there\u0027s\nno need to specify it separate from PBR in build backend settings. It\nshould \"just work\" everywhere.\n\nChange-Id: I015b271335b9419ac5a73f7f19320087dd5eb1a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d4a1815afa920cf20e889be20617105446f7ce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d4a1815afa920cf20e889be20617105446f7ce2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
