)]}'
{"id":"openstack%2Fpbr~948756","triplet_id":"openstack%2Fpbr~master~Ib452a08c577913808c1ded67240b2101dbb7f7b5","project":"openstack/pbr","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-04 13:03:58.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ib452a08c577913808c1ded67240b2101dbb7f7b5","subject":"DNM: trying to workaround new setuptools","status":"ABANDONED","created":"2025-05-04 08:40:40.000000000","updated":"2025-05-05 08:19:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d555b2a97cdbc0cac291c434a3a7bf95c5647868","_number":948756,"virtual_id_number":948756,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-04 08:59:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-04 08:59:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-04 14:35:08.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"CC"}],"messages":[{"id":"fe7717075891af9d0ad2113a6db5e89253c9cb3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-04 08:40:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5600883eaf3c1315fe01da0a45d5faa1c98db970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 08:59:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/333c61cad8c442c58283acc79e64d0d9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c39a18faaabf402f8bf51c602aad3bd1 : FAILURE in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25eb5a286f7a4456b4c019866c17971d : FAILURE in 2m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a503f3e233343ee8efc88eb6de7d28a : FAILURE in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5248244cf5fa400e963f07c71269a21d : FAILURE in 2m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d306d258c8f4d0183cb2d68048f16d3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a793b6c768424328971ca393cfdd9e76 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a07c6458305a4bd7829d1b8063d05f3f : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f6c86ae9e1f4fe0a8a010bdaf89a849 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1a5cede499640e8a5e96ddcfe5329f5 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69f07aefab964db790a37a6bde2b2828 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a19e2e087dd04ec59ebf10a790dabdb3 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/133731a0c5ea4843bd4c5db3530ab460 : FAILURE in 1m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ccf30896370f409fbdcc829ab782f244 : FAILURE in 3m 46s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/0c687b25c8a04fee97d30bbffeba2a8b : FAILURE in 7m 46s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/66a24f6b107f47d4a88757ea880a8726 : FAILURE in 5m 23s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/2efa30d718ce4cfea65502e1d61f7548 : FAILURE in 16m 47s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/1cf1cb892eff43d9b6deeed63eed5835 : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d555b2a97cdbc0cac291c434a3a7bf95c5647868","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-05 08:19:21.000000000","message":"Abandoned\n\nsetuptools 80.3.0 has been yanked","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ceeb9c7c0f638711a8bf0c195a946d5cc1ca826","revisions":{"8ceeb9c7c0f638711a8bf0c195a946d5cc1ca826":{"kind":"REWORK","_number":1,"created":"2025-05-04 08:40:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/56/948756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/56/948756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/948756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/948756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/56/948756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/56/948756/1"}}},"commit":{"parents":[{"commit":"a4ee66bf8c68a2cd394e590c6bc155068d208d25","subject":"Add openstack-tox-py313 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a4ee66bf8c68a2cd394e590c6bc155068d208d25"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-04 08:37:45.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-05-04 08:40:07.000000000","tz":120},"subject":"DNM: trying to workaround new setuptools","message":"DNM: trying to workaround new setuptools\n\nChange-Id: Ib452a08c577913808c1ded67240b2101dbb7f7b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8ceeb9c7c0f638711a8bf0c195a946d5cc1ca826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8ceeb9c7c0f638711a8bf0c195a946d5cc1ca826"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
