)]}'
{"id":"openstack%2Fpbr~949051","triplet_id":"openstack%2Fpbr~master~Iff8d3de2358fc9f66abcb659d864cb4f3df40625","project":"openstack/pbr","branch":"master","topic":"pkg_resources-removal","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-04 20:03:35.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-04 21:40:08.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-08-04 21:40:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iff8d3de2358fc9f66abcb659d864cb4f3df40625","subject":"docs: Update usage guide and README","status":"MERGED","created":"2025-05-07 16:30:11.000000000","updated":"2025-08-04 21:41:09.000000000","submitted":"2025-08-04 21:40:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949051-pkg_resources-removal","meta_rev_id":"52c2c56fe96075e6cafd620f7328a6d31f526522","_number":949051,"virtual_id_number":949051,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 21:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-07-17 09:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-21 08:36:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":2,"date":"2025-07-17 09:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-21 08:36:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 17:02:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 17:27:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-07-08 22:50:33.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-07-16 15:58:03.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-21 08:36:12.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"043984cd1e46ad9d7cc2c3dae92a924603282ba9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-07 16:30:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"385c509c9c7ffb9101ad0d3abc8e6a3fba8f858d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 17:02:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d80cdf16a924724b5d8e496df197121\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfa6075cee94473388e4173a4c6bfa50 : FAILURE in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36cf74b794a4a9eab19e46dc21ced56 : SUCCESS in 2m 25s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/3d5ae294aa924242afd0d234c258106f : FAILURE in 22m 00s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/ec827358fc554e99ae880066834d26e3 : FAILURE in 30m 56s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/85f1bb4862c24888adf27028412a75d9 : FAILURE in 15m 37s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/18729d7e0b7240e59358ecfbfbe497ce : FAILURE in 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c61a01e58fead436048efc545de597a2faef0144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-08 13:29:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8beed71a19adeff059c87a51613383142878daca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 14:00:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afa4e71c268c497480d5a47dcc71735f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f7f0b0281f34ef8a2f7d0f801a49e5f : FAILURE in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0845848b6d44dc7bebf337543f6098b : SUCCESS in 3m 13s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/857ff7861b8546caa69b6b7eb612e3bf : FAILURE in 28m 38s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/57cd5b0e87664b73bdc2f105a34a8f63 : FAILURE in 23m 20s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0ebdbbb1b8d449939fa0e89eec43da25 : FAILURE in 16m 00s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ce70c726823e419f9a5cf9d91f0d214f : FAILURE in 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1a549af0d1b0520c1516dddfe34ed9dd90739fdb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-08 17:27:49.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dbf066850948697af5ccc31da4d2c729a1182212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 10:30:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5d6dc0c939b9cb68167e6156e6c747d9580d5ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 11:17:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e1e042860934ea781f826383d6fc15a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/788153a07f044bde909fa91bf98791e7 : FAILURE in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89495344b9848bc9f156f94f47107ed : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe9018fe71346a5badee1c05c6ab0e0 : SUCCESS in 3m 08s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/85a3bfeb8053409c98ede3a7f09b17bc : SUCCESS in 24m 49s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6aa4cfd3524f40488e9872eea6dcdce6 : SUCCESS in 29m 32s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/905c5d5d7ef045a798950cb8942e1ca4 : SUCCESS in 41m 15s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/bf566cc2ffcc4514b5c091bd54cba0c3 : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":3},{"id":"83df0ddc9500a86a460791219cb965aac39a8f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:34:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7e35829de2fab29b36e0a15833a516b545603351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:37:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0368357a8213d691b33e91ad7c767faaf98c315f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:37:38.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cc51e715e8f626f3e22c6b0f7d6b332041c36042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:40:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3e6763a9cd5e3e5dcfddf67fdcf5ad14aae9752f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:21:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12ab373a303540f99b85bba686e45e4f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e72e18407fe4050939892e14cffcebf : FAILURE in 1m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e83f4008a2d44e6877226e314cb0ba8 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b577153c8bc8424a826e775f9a2c4d02 : SUCCESS in 3m 30s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/fba937cd244d46dc91d59f610a845871 : FAILURE in 8m 19s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/dc88fff743ee4e099149618e135bf8a7 : FAILURE in 5m 22s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4c31c15645ee49b599c1aa97303bfe55 : FAILURE in 18m 13s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b31f667899494f609ffc2789e787d615 : FAILURE in 15m 42s","accounts_in_message":[],"_revision_number":6},{"id":"f6d832fbc1b20bad4b1affdb23165b98a0d93072","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:13:15.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ba11e7807f9496fa2bcca1f77d927c52440e0b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 09:54:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0676b7282973404f86a1e2af5073c513\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/978eb8c258e9445c8ba2975e4a7110ac : FAILURE in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483e2521e3db422c9777ae9ffa7e04c5 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9c52f44f3d409f88f42f33c2141f69 : SUCCESS in 2m 40s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/bf5c7d37a5544da085f76d84335601b3 : FAILURE in 28m 08s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4247eefc79bd4f8cb1fac311a6291ae9 : FAILURE in 29m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/065af499579849cc94a5d50038c6fde4 : FAILURE in 21m 55s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/8d1774f7c83e4e2193412caa671614e6 : FAILURE in 20m 05s","accounts_in_message":[],"_revision_number":7},{"id":"a5ead2876bb2195cf6d4981e46117a67207501a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 11:00:37.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d9ded869ac47bd665ba95d8fec51bdb8688f7f85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 11:45:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07ec737d55cb474e9282e768345c6954\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/14b20d8913a444658c547e356d943111 : FAILURE in 1m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbddd65e478e413d9cc99af3254a47b2 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1992aac20e4fbdb9cad886dcdacfa9 : SUCCESS in 3m 03s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/c63733721e404ecaba6c6d2a9a3a9dfe : FAILURE in 24m 55s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/79086066966c40999981a1ce2c9526c8 : FAILURE in 30m 44s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4bec5851b9a949b490322ac7a43bfc24 : FAILURE in 16m 46s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/da6cfbb5bc604b77aee52783d64c92bb : FAILURE in 27m 50s","accounts_in_message":[],"_revision_number":8},{"id":"d7ced518d66b787e99a542c69ffe4ec681eb80d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 17:59:08.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d3c4d7647bfcf390d5bd5c3965f6c8e0d413e148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 18:33:15.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04be52857e8c42aeb660f5c394173244\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c38b7d871324e20af373be49220d1a8 : FAILURE in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606da14fa6914d039058616ad2dbce70 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de6bbc3f66e9485bbc9abf2d75916b7a : SUCCESS in 1m 50s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/648a490b1be642db9fc8bb8640a15a39 : FAILURE in 24m 50s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/cec2afb3772646f5b5663f1466984b15 : FAILURE in 26m 07s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0fe6cb8580b3471d880577edaf51a4f0 : FAILURE in 17m 26s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/7b79146e9540438985add79d94a52725 : FAILURE in 18m 03s","accounts_in_message":[],"_revision_number":9},{"id":"ed61d913f0e88c29c783899b31f332f7f3a4a179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 20:35:45.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"7676392c29b0e3bf9948ca3fe0c63e135b6bc4ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 21:15:26.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/067a2b9cefa240c990f8e3b8ff71794a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0bbf65dd3b044b8b9053daa55c9ab789 : FAILURE in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832efc5866144932aef1bb36a65961da : SUCCESS in 1m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0519c47dc23b44e3b4dad0fbe6c7d98f : SUCCESS in 2m 06s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/163f8c8178404a26ad660a771d09ba01 : FAILURE in 14m 03s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/5a6bd485041f419da9aa38e6cf2247fb : FAILURE in 33m 21s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c7ca6e2a4c414d5fb8ae4d5aba094e6d : FAILURE in 16m 11s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0b31950c51f34305a0946038cb9cca0e : FAILURE in 31m 58s","accounts_in_message":[],"_revision_number":10},{"id":"2ef5ca3effa91583edf8d63404846a92b3a7fb10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-05 17:42:22.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"cf896261f78d8e192315877ed76cb9d45a975f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 18:56:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d40a63842704c8481b85c791b5f2d26\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec188a0d626e43b0aad7f13b4dcbaf97 : FAILURE in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27bafec246254e2aa976ff67a25a2c5a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a62e82b9634733800a61e802fad975 : SUCCESS in 2m 36s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ebe8482282454311881e7ddcc6e13979 : SUCCESS in 26m 01s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/5a8fe198285947d1958f940a09d7e6bf : SUCCESS in 28m 49s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/332f3e55788141619ead8467f7c4482a : SUCCESS in 42m 03s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/60ea63c5b308459cbbfb595a94ed905c : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":11},{"id":"e9ec84c358c0c6a75107e4b35658fa32a9c37bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-07 08:30:36.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"a2d95eb3f95bf75dce85091071b23e77895f680e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 09:02:18.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/189644b0520d4a34bf9321ec6655a3c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c10233b8b5d450e86baf629dbe13a75 : SUCCESS in 2m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76db43261b24f03bfa2283ab61e24d7 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07502c486b6344b38593b0d87c6094af : SUCCESS in 2m 56s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/b6c495c9c10147ea93df3e23e78871fa : SUCCESS in 13m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/c6c5f13a687a499ab85c973506889aa3 : SUCCESS in 27m 00s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0974d3a5726649fabb588f74c253f24e : SUCCESS in 22m 04s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/fc7d5830dd534f239e189690e3bc71b6 : SUCCESS in 17m 13s","accounts_in_message":[],"_revision_number":12},{"id":"a51102a0fa33a8c07a4168ebc2629b7e48e8e9d1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-08 22:50:33.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"ab4d2bc13df85f1c71369fae675db5df3fb14931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 15:04:16.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"ddcf6baaabe35497212d1fbead34c1412b364a88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 15:55:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96a83740f9fd4c7c85d0754e704935d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a24b43ca1674d2a9b1e18d34ec401ec : SUCCESS in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c371cea31114e31bbe7937517399722 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a3cafe33b742e1938af1c9fdf96358 : SUCCESS in 2m 06s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/31b6b7170c4245faa836c47300eeff26 : SUCCESS in 32m 34s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/69a1cce06d084bc58553dadbf2ac7455 : SUCCESS in 22m 49s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/af77cbbe76404d9c8e7d24da6296315e : SUCCESS in 30m 56s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/67ed1e0428c440078080c223a3b2ebef : SUCCESS in 36m 50s","accounts_in_message":[],"_revision_number":13},{"id":"137a2faeca1ced1edcc96faff315f54eb908daf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 17:03:50.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"83edaf0e46d598e3defa9b145d23cd8c642b0836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 17:48:23.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f032db19301542ca91328597c7d68d41\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a17a1fe78fce42be8bf07e34a2c73ca1 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8d34d650654c0987226ed200df7b1a : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fccef708a9c2421a8852d7c77345941c : SUCCESS in 2m 34s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/8e0deafdff3e4baa94b4d71d7c8bd25a : SUCCESS in 23m 21s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/c7958df7045444538ad78ae37ce91dec : SUCCESS in 29m 42s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/31942ca9d1984930b9ab0e14b3b40273 : SUCCESS in 33m 56s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/9d4de567be05494d8792e780a4e090cf : SUCCESS in 39m 26s","accounts_in_message":[],"_revision_number":14},{"id":"42768f10ba66614fbc6d4d9d50e3c1b0d66f9233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 10:37:26.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"1c03948f47a510f39d3ba856316c020d9c488db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 12:12:22.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f593dcd2cb664592942f5f55c068db80\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ca195b9da2ea471988e71c79c27ac9ef : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9df8988b703a4ffcaa45b585eb09eed5 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94049646f8f84544874728f58b7315d2 : SUCCESS in 1m 51s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/414b18e072ae4819a3e61560799932af : SUCCESS in 21m 40s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/9b42e29f50514668a93e53a696079ef2 : SUCCESS in 24m 23s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e81f24ab6e854167880f0077717a32e2 : SUCCESS in 39m 26s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/7a7846dc5d0d4744ba75508c6f2f4264 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":15},{"id":"99dd4067190b0ca7178edfc7327eae6f991254c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 12:46:27.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"5853b2c44b2b861e622b320e055d8336e70b2385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 13:44:43.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fec566e852eb47c985e520aa9b78abbf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab913330f42c4939928f5823381b77d2 : SUCCESS in 1m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6344de53044c868144c6a998627360 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a1b790da5d4214a665b48998c70415 : SUCCESS in 4m 07s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/9b91e3abd6ce4d12944c715aa9fc572f : FAILURE in 15m 34s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b4fb4ecaf80a4acca5e496254f6fb390 : SUCCESS in 29m 42s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/501e937988ec4bdca4bd291a732bdb9c : SUCCESS in 42m 33s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4dd01a957a9c4be5a0fcdff396d12200 : SUCCESS in 40m 51s","accounts_in_message":[],"_revision_number":16},{"id":"c8fca09054afeaa33c90cc4a7b58f4efd382a21d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-16 15:58:03.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"2773adb1071a5dae53351ad08bdcefae21836895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 15:58:23.000000000","message":"Patch Set 16:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/pbr for 954049,11","accounts_in_message":[],"_revision_number":16},{"id":"dabb43c5b63d88934a7aa2151e4cf3c3203fd73a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 09:32:27.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":17},{"id":"e6ea72a49a4b5a63c96c77961b6feca85d8731d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 10:19:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d28eefdbfcc54fc5a1764878748434d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/86550c04bbdd46168479813f9f692c13 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de466dbc6c841b58084b650758491b5 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060b56144a9d4cd2b43b95154079aa19 : SUCCESS in 2m 18s\n- build-python-release https://zuul.opendev.org/t/openstack/build/944e0638005b47b5b6b3813dcbc0ad30 : SUCCESS in 4m 03s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/68601230e4cf4222b9897fac0128978d : SUCCESS in 23m 37s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/c140fd5eb2d5419d90c158225405b5f1 : SUCCESS in 27m 44s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e2477f1f95db4bac98d62050c349dbe8 : SUCCESS in 31m 08s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/528efb35ade84238b94ceec891b5c99f : SUCCESS in 41m 44s","accounts_in_message":[],"_revision_number":17},{"id":"61c67997f4841305612ce2fa4eda1d9a5db78dac","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-07-21 08:36:12.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"8978f368f8a5febcc009ba61fbd02787a9f09bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 16:35:00.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"ff8380781edebdd0924f20be0a6bf9eca998cdb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:05:31.000000000","message":"Patch Set 17: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/954049 is needed.","accounts_in_message":[],"_revision_number":17},{"id":"3833652b216844b4f03caeb6d855b2787264179d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-04 20:03:35.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1952b27a97ab60edf780d8a09d99c9491bdcb8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 20:54:11.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2792981ae6ae44c1929bc8e3cd3b1f52\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5c156974126645ed8de416d55d244cae : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f14936f3771467986357f061447bb26 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9545b0517940409552ce8f2502b848 : SUCCESS in 2m 34s\n- build-python-release https://zuul.opendev.org/t/openstack/build/ecabebd0e0a54d30b15f075758ad1621 : SUCCESS in 1m 57s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/5786ed31043440eab5f1b8df32850eb7 : SUCCESS in 27m 43s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/017006e3bf8747b9bdc1ebfac969dcad : SUCCESS in 26m 33s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/1adac61d7835405d986289eef4b33932 : SUCCESS in 38m 04s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/414d70ea645e4171a95218bfd332076a : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":17},{"id":"4e7ae8145e6b8ce889f5b389da797c600f55c097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 20:55:20.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"47b5b401802dffa16798cef52d0062061fc934d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:40:08.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8476cfd0f57244a48247bb7be7043b23\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad78f0d25e3a4f17b9176332c4012a08 : SUCCESS in 1m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06de6f05d510425abeaa414daf59b160 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9c2bc4acd24010ba40698133c49987 : SUCCESS in 2m 49s\n- build-python-release https://zuul.opendev.org/t/openstack/build/3bd45d1ac8b3405bb4930c8200398a9f : SUCCESS in 2m 09s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/113526076bd542ef9f03f2ff6953058b : SUCCESS in 16m 45s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d2e25ac16c8a4e3f871e7680afa79781 : SUCCESS in 16m 53s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/15dd5883616048f1aba746ce69280f25 : SUCCESS in 43m 46s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4d0dc1c26de44073b5391c66870c49f7 : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":17},{"id":"f8ceced5c85ffe8388a8075842d50a06d8f4e655","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:40:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"52c2c56fe96075e6cafd620f7328a6d31f526522","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:41:09.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5aaee41712e4012a474f7108567f3a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82fc827209a344cfbe7e3aac3df5213f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"8d5f4eb020334f10c48e391e6b8c719f35104444","revisions":{"2f1d19c5034d08b150a7d0b58e22e6e1db5540f4":{"kind":"REWORK","_number":1,"created":"2025-05-07 16:30:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/1"}}},"commit":{"parents":[{"commit":"a28ebf605b1976581135725e22847e39595f094c","subject":"docs: Reflect changes to setup.cfg requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a28ebf605b1976581135725e22847e39595f094c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 14:46:12.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:28:25.000000000","tz":60},"subject":"docs: Use code-block for syntax highlighting","message":"docs: Use code-block for syntax highlighting\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2f1d19c5034d08b150a7d0b58e22e6e1db5540f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2f1d19c5034d08b150a7d0b58e22e6e1db5540f4"}]},"branch":"refs/heads/master"},"e9a5c9401e00cf0bfdda70fb2311796fa23f881a":{"kind":"REWORK","_number":2,"created":"2025-05-08 13:29:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/2"}}},"commit":{"parents":[{"commit":"13392c45b32bb679b68f228f33617051b363824e","subject":"docs: Reflect changes to setup.cfg requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/13392c45b32bb679b68f228f33617051b363824e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 14:46:12.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:28:58.000000000","tz":60},"subject":"docs: Use code-block for syntax highlighting","message":"docs: Use code-block for syntax highlighting\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e9a5c9401e00cf0bfdda70fb2311796fa23f881a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e9a5c9401e00cf0bfdda70fb2311796fa23f881a"}]},"branch":"refs/heads/master"},"4f9aa9601ede6bda3bd01a773b8637a5e2c652f1":{"kind":"REWORK","_number":3,"created":"2025-07-03 10:30:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/3"}}},"commit":{"parents":[{"commit":"db65d0b5fa09f7f66490def0c55d3c28bc4232a7","subject":"tox: Stop calling \u0027setup.py\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/db65d0b5fa09f7f66490def0c55d3c28bc4232a7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 14:08:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 10:30:16.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f9aa9601ede6bda3bd01a773b8637a5e2c652f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f9aa9601ede6bda3bd01a773b8637a5e2c652f1"}]},"branch":"refs/heads/master"},"94d667a2705ba2a197671814758dbd69b0086408":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-03 14:34:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/4"}}},"commit":{"parents":[{"commit":"abb57b6436403f0d0febe2622daac4003dfba44c","subject":"tox: Stop calling \u0027setup.py\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/abb57b6436403f0d0febe2622daac4003dfba44c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 14:08:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:34:14.000000000","tz":0},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/94d667a2705ba2a197671814758dbd69b0086408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/94d667a2705ba2a197671814758dbd69b0086408"}]},"branch":"refs/heads/master","description":"Rebase"},"4a58993da253cc33578ee698f85dc96ad1ce9236":{"kind":"REWORK","_number":5,"created":"2025-07-03 14:37:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/5"}}},"commit":{"parents":[{"commit":"abb57b6436403f0d0febe2622daac4003dfba44c","subject":"tox: Stop calling \u0027setup.py\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/abb57b6436403f0d0febe2622daac4003dfba44c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 14:08:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:37:14.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a58993da253cc33578ee698f85dc96ad1ce9236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4a58993da253cc33578ee698f85dc96ad1ce9236"}]},"branch":"refs/heads/master"},"d978905564b224a3df653c398e011a7e61606f9a":{"kind":"REWORK","_number":6,"created":"2025-07-03 14:40:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/6"}}},"commit":{"parents":[{"commit":"6731d6b5a7bf826d61226bde15df597967908bb6","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6731d6b5a7bf826d61226bde15df597967908bb6"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:39:18.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d978905564b224a3df653c398e011a7e61606f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d978905564b224a3df653c398e011a7e61606f9a"}]},"branch":"refs/heads/master"},"703f7887ff8b19ef27592cecaa06fe415263ea51":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-04 09:13:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/7"}}},"commit":{"parents":[{"commit":"8a4c6a5370da7f076ee9a172c919a4b200723d7a","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8a4c6a5370da7f076ee9a172c919a4b200723d7a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 09:10:35.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/703f7887ff8b19ef27592cecaa06fe415263ea51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/703f7887ff8b19ef27592cecaa06fe415263ea51"}]},"branch":"refs/heads/master"},"3b79e4156800eb4997122a82c4bbb3dbca30c531":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-07-04 11:00:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/8"}}},"commit":{"parents":[{"commit":"ab008b40adb4e0ed62d58b057b43e25bd04c9104","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ab008b40adb4e0ed62d58b057b43e25bd04c9104"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 11:00:30.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3b79e4156800eb4997122a82c4bbb3dbca30c531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3b79e4156800eb4997122a82c4bbb3dbca30c531"}]},"branch":"refs/heads/master"},"a9282bf7162a1e69f8f12e470c1c0324f7d47e9d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-07-04 17:59:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/9"}}},"commit":{"parents":[{"commit":"65970c9a093789d3b3cb951ee26ebfeea966fc2f","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/65970c9a093789d3b3cb951ee26ebfeea966fc2f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 17:59:02.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a9282bf7162a1e69f8f12e470c1c0324f7d47e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a9282bf7162a1e69f8f12e470c1c0324f7d47e9d"}]},"branch":"refs/heads/master"},"bb80d9e8cf6361f2f6732e7d81a2e42d120a54e4":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-07-04 20:35:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/10"}}},"commit":{"parents":[{"commit":"2ce3a44beecaa857e301141b2e7b066a43c53c5e","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2ce3a44beecaa857e301141b2e7b066a43c53c5e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 20:28:47.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bb80d9e8cf6361f2f6732e7d81a2e42d120a54e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bb80d9e8cf6361f2f6732e7d81a2e42d120a54e4"}]},"branch":"refs/heads/master"},"8d78fe20cf434ac463b296c63b321613643b8a0b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-07-05 17:42:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/11"}}},"commit":{"parents":[{"commit":"812463c4f446a0ea3dba91d38919bd992b76cdf1","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/812463c4f446a0ea3dba91d38919bd992b76cdf1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-05 17:42:04.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d78fe20cf434ac463b296c63b321613643b8a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d78fe20cf434ac463b296c63b321613643b8a0b"}]},"branch":"refs/heads/master"},"2828eff204186cdfc1a54af307ececfa5aba7a06":{"kind":"REWORK","_number":12,"created":"2025-07-07 08:30:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/12"}}},"commit":{"parents":[{"commit":"812463c4f446a0ea3dba91d38919bd992b76cdf1","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/812463c4f446a0ea3dba91d38919bd992b76cdf1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-07 08:30:15.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2828eff204186cdfc1a54af307ececfa5aba7a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2828eff204186cdfc1a54af307ececfa5aba7a06"}]},"branch":"refs/heads/master"},"31c2fb6a3f57a1276c7f214990a926a3f89c2cff":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-07-09 15:04:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/13"}}},"commit":{"parents":[{"commit":"e5449263ca63c4690496e1247bd63877988621c5","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e5449263ca63c4690496e1247bd63877988621c5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 14:27:31.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/31c2fb6a3f57a1276c7f214990a926a3f89c2cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/31c2fb6a3f57a1276c7f214990a926a3f89c2cff"}]},"branch":"refs/heads/master"},"68d05f55c426894bc2916249be855c0730644915":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-07-09 17:03:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/14"}}},"commit":{"parents":[{"commit":"fe1b18a46eabee78572ca017dfd7d61f74781dd2","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fe1b18a46eabee78572ca017dfd7d61f74781dd2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 17:03:38.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/68d05f55c426894bc2916249be855c0730644915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/68d05f55c426894bc2916249be855c0730644915"}]},"branch":"refs/heads/master"},"8b0e6d7262862e6a4e8c5838c89a91397c6b06dd":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-07-10 10:37:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/15"}}},"commit":{"parents":[{"commit":"8cbf7c94f5ff8ad80cf4e802c7f32ad42f0cda73","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8cbf7c94f5ff8ad80cf4e802c7f32ad42f0cda73"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 10:35:29.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8b0e6d7262862e6a4e8c5838c89a91397c6b06dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8b0e6d7262862e6a4e8c5838c89a91397c6b06dd"}]},"branch":"refs/heads/master"},"b0e7bef5ebd5b222c8f4f9712827b726539c3893":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-07-10 12:46:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/16"}}},"commit":{"parents":[{"commit":"3addcdb4834217ed5ef156ef3514d65bde60c789","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3addcdb4834217ed5ef156ef3514d65bde60c789"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 12:41:23.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0e7bef5ebd5b222c8f4f9712827b726539c3893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b0e7bef5ebd5b222c8f4f9712827b726539c3893"}]},"branch":"refs/heads/master"},"8d5f4eb020334f10c48e391e6b8c719f35104444":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-07-17 09:32:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/949051/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/51/949051/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/51/949051/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/51/949051/17"}}},"commit":{"parents":[{"commit":"6dc031c57233fa919cfcb9b5c7dc528a248fb404","subject":"Move functional tests to their own directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6dc031c57233fa919cfcb9b5c7dc528a248fb404"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 09:32:05.000000000","tz":60},"subject":"docs: Update usage guide and README","message":"docs: Update usage guide and README\n\nBetter reflect the purpose of pbr. Also make use of code-blocks and\nupdate some examples.\n\nChange-Id: Iff8d3de2358fc9f66abcb659d864cb4f3df40625\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d5f4eb020334f10c48e391e6b8c719f35104444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d5f4eb020334f10c48e391e6b8c719f35104444"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
