)]}'
{"id":"openstack%2Fpbr~949053","triplet_id":"openstack%2Fpbr~master~I96e4cc5f7697bbae398bf496b0111781a1d1708c","project":"openstack/pbr","branch":"master","topic":"pkg_resources-removal","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-04 23:41:24.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-08-04 23:41:24.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-17 17:12:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I96e4cc5f7697bbae398bf496b0111781a1d1708c","subject":"Deprecate \u0027[files]\u0027 section","status":"MERGED","created":"2025-05-07 16:30:11.000000000","updated":"2025-08-04 23:42:07.000000000","submitted":"2025-08-04 23:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"949053-pkg_resources-removal","meta_rev_id":"d4e1eb1120cafee30115f0c5e0d9c7e9b90666b8","_number":949053,"virtual_id_number":949053,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 23:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-08-01 16:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":2,"date":"2025-07-17 17:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-08-01 16:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 17:38:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-04 11:13:07.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-07-16 16:07:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-08-01 16:39:16.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-08-04 20:04:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"85731c5d6d664451b8bdbf5a3651a1275c6e0a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-07 16:30:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd2dc1dfcb09031c58ee89911c4717403efced14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 17:38:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75199b9519cc4b1792eee2bd9aef8767\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1fb11fc180e040f181032facbec0ced0 : FAILURE in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50936ce197db483689cd96acd1217784 : SUCCESS in 1h 05m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052945c22dd1473ab3282d09a5caff34 : FAILURE in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf30f70732242709d11f6658b031061 : SUCCESS in 4m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b32ab3224a74deeb10a844fd66c0021 : FAILURE in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea69665d8e6a4dbc9f371146eceb93b6 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c27c262fca5f498fbdbbd6f8faa9b44a : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61aaadf5191646759ff067e9a5d2f707 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4300b53ca6f64da4922f627af6e72d19 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/def3635bd0e14a4b822d5f7a9149ce88 : FAILURE in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6e93d2e1ac746a2903865b319a7f834 : FAILURE in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84dbab00a4dd4f809b2549070d96de92 : FAILURE in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27937195430e4f9dae909339b025ee50 : FAILURE in 5m 46s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/4f67c89519e94dc3b4ea09e251ed32b0 : FAILURE in 18m 56s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/a068ff474cb04fcbb6605dd14c4419da : FAILURE in 22m 02s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6bceb0dfc9ab40e7a87e4b24e988dd18 : FAILURE in 16m 39s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c01569602f2a47be8e4a7c39671837d1 : FAILURE in 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"090102cfee46d19cfc688180b7405a0a8805f7e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-08 13:29:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fc2bac4243897a6f7ead4411520b1c9a3f19a198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:17:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4125ccf52c514e70a6856afd1c27c53f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/faa1a464d66c44008247aa045c032a5e : FAILURE in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87600fc2a7a6453a89bb211603de5573 : SUCCESS in 1h 31m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92fb52f23234395bd92108e70d13fe2 : FAILURE in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8cbacda238d44299a5821f24221ccb7 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a3a793a1ed54f12bbb4033de4e37a53 : FAILURE in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edf2ca5c31354214af6daa24d6882128 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0591042ceaf14063a38a29c028817372 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8af753f2ae5e402e8bfed8add4a7f8a4 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c465734440c14c1ab21ef315643a6ff9 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/142cec58bbef4f6daa69222f1cf33981 : FAILURE in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/248a6824980e4aafba1502fb5bf16414 : FAILURE in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/612f9f4c16c04642bc516e8deab8f031 : FAILURE in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1e6597f3f3f462bbe25bec2b4accd2b : FAILURE in 6m 42s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/32cd1131e9594df59a702144adb8ec1b : FAILURE in 28m 48s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/572219b16be74a1a976bcf95f8b20d3c : FAILURE in 13m 53s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/5a96420e3ba442a58f0592cb31e3e298 : FAILURE in 15m 43s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b6b971dd4b6c461dbe3661ddfbc40047 : FAILURE in 20m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b1304659c2489a546fd267878c1eca204b140438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 10:30:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3e0b3961a30c338f8185b8aa9045878d2f3e7bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 12:24:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f29c332cf74b4b2a9846c30bede8afb1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cec7520c0eee4aadaddd0708c10bde9e : FAILURE in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c543a9ad25754416afbedb7d7ddc4ed9 : SUCCESS in 1h 40m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed092f53271043bfa792e2d576669720 : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56afe8a1c6045e5a4ca33335134bb9f : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be689da1ffad4a699f2fae14c0ff7203 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ad54dcd333d4d7d930179f74b7746d0 : FAILURE in 1m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3d7148a8dc74824b0cb934cdf65ad46 : FAILURE in 2m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea56f14fbe214b71a0f4ddac40e49337 : FAILURE in 1m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/837d3ddc976d4e268ca8be14bc5086b6 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f303b9f3910a48cdaf43d172d0a699b3 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68ffe192670a4561ac687bfebc080597 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64282071de3f4641ba233bda63a9f728 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b923ac2ddb024d5fbe6196d76164002a : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a742c3f0a09c423babba31aba3565ec5 : SUCCESS in 7m 50s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/fe06c244a1bc4cc7b2c0fec0c87369cb : SUCCESS in 29m 20s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/a63c2d200cc149a48686273adffd16fe : SUCCESS in 26m 42s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/b68e5f3ac105443ba04a92d840718f5c : SUCCESS in 31m 51s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/67619d0257e54b2dbc7cebbc9bfcc189 : SUCCESS in 31m 29s","accounts_in_message":[],"_revision_number":3},{"id":"aa09c7a5562fb6741e32ce9d4bb313c269e5867f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:37:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"efbc3c1133e3b3e7b8b26d423a5bfcd6799f5d68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:40:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"fefe34c314eaceebc8d5c68a34b19d900cc9a660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 16:32:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/901fbd303e6347159b889513ab35aad2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b0e8a12a25414335ad6140975d855785 : FAILURE in 2m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f010910cf2b4287bd3b60555affa42e : SUCCESS in 1h 43m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270eb83193844c638276ab242f09340f : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed12f12b6b94e2dbc4232c6b9bcd14d : FAILURE in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/944f3725da72410fac47232d1de4ffdf : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0058c7e91dfc46e5a36b32c78fad7ed4 : FAILURE in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22b5f3280992460b95a154cf71a384be : FAILURE in 1m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfc66595199e498e83f881d468ce63c6 : FAILURE in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/472f898e93204a7b9543882cd48330e0 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/021fd491dd0c4187b39684dd8f7f4903 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cd29760df814630bf24a3be48859f67 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b5bc230478142548f61d6d30a05f631 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b4d581c2b294e6eb8aea206cf34fe1e : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c240cf5d480443da9543f7868652f44 : SUCCESS in 7m 29s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/a16f5d33742e47988b699b0c4de7e4b6 : FAILURE in 4m 29s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/3e2802953da2467782c8fb85949162a4 : FAILURE in 9m 17s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e7f36f8479404a23b316711d7484e6c8 : FAILURE in 15m 09s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c1c45690f33b4e4e953090c8ac69f59a : FAILURE in 21m 41s","accounts_in_message":[],"_revision_number":5},{"id":"041fece98570ff2a1f6774e45a62b5fcb88beb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:13:15.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"52fd15c2dfa464e934e3e586445788de48cdd194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:59:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27ff5481a88e4b268d94811866e143a3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cc59094d770d4c11b2b57af36dcd287b : FAILURE in 2m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/163091fc202b43bc9acce46b8374be0a : SUCCESS in 1h 38m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea2ee4a6d3948cda034d495c5208d56 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1edee70a728d43fca930bf42505e9642 : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42507a58d42344f0a0f35f2203c13efc : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78fb9c87b0ff4b47a99fda3f600bd0a8 : FAILURE in 2m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9df9ad18ac364d128b1a28a27096bff2 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cbbdce12d6d4aa48f8499a38361d885 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e042ad52ad64082aae11eebe2fcfcad : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5041faa61b7a4b81afe1e9dd5513c1b1 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63acbab080d34369aadec1ee700c2665 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/766658799ce14a9ca41757df75cfaf79 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8871a799270421e874af798e0d5fbd5 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df03c082ea794121adf97a3db8174f4c : SUCCESS in 8m 44s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/773ad4f606f341e8a4453d2cd2ad0a59 : FAILURE in 20m 05s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/206b2a81c2a84b8488fc78076090d0cb : FAILURE in 21m 53s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/70d6749a0a894f458b95726d471e101a : FAILURE in 35m 03s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ee601d1b3ead44b1aa7362d861e2757b : FAILURE in 27m 18s","accounts_in_message":[],"_revision_number":6},{"id":"5c1975cc93e466a154968bd24c13f8b72d396c50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 11:00:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"3b4568f8fc53956478b7d618e955385776951761","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-07-04 11:13:07.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"56e2e3ed34cc79010a4c4c7d047f0251fa7be743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 12:47:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fb38c5a84c2403c89a00c0d53807182\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ebea853e4bbd4f76a0c3208f79eec3ae : FAILURE in 2m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8053a7307723414b9cfc883363f58a7a : SUCCESS in 1h 39m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef49344d53d4bbf81f9d8f8f1ab256d : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713944a859cc40ad92b9a08584c65c71 : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/634f67b4e4884c7f98ba8173aa13977b : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8c8804a668b421584739918a5bd7b2c : FAILURE in 1m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b468cb20928b4023be0e7ffc54d47942 : SUCCESS in 2m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6b9649cebf8431ea1cfb4ac83545d7c : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdb55559aa204f4988eaaf3e3859fc17 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13c6f972edbe41dea1bd518a33c6af79 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9178fc8e48b41b49b1e2a7f1896a428 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bd09d4eb0484720b5b1b3ec3ba23f8d : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b574864b02314b719a536c4c80c02906 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f458cda0e1cf405795f32cdfb7d6a32d : SUCCESS in 8m 11s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e11733eaed01474aa591714ca296b61a : FAILURE in 24m 46s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2928fea4d74a4d60aaa53d5c18a27d22 : FAILURE in 21m 59s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c6e0583a4d8343ca91c2ce18deecbf52 : FAILURE in 29m 27s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/493596f4725e4746bf1f2f66f809906c : FAILURE in 30m 08s","accounts_in_message":[],"_revision_number":7},{"id":"6bb7990976f3fccb7a46474845069f275ec87099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 17:59:08.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2c51d4309429f130e7e9482f86f14e4b17c52769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 19:59:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/593727328f7244c5b1fd58d91290e273\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/116f91fd2dac42199ef523dd9ae3b3cf : FAILURE in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3408ac548604083b55486ea76daa06b : SUCCESS in 1h 26m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f5f789586e4620a6ad2b0081e468d9 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79510c991be4459a71a3a76bdd969cd : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5db65cfe57e84cefb43da2f309d242eb : FAILURE in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9aebaa4ff6b04a3592160775cd8e1e76 : FAILURE in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/864fa84a3dc54f1bb1ac79e114e0f12c : FAILURE in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/600e892021444f61bb35aa8a0f582e4c : FAILURE in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bda37c6b09aa4c3097f66e9a76a0c1ab : FAILURE in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48cc6e07d3b6468dbfbad6d4e0c17a5a : FAILURE in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ca260e2901f477c9d46272e34f6f7fb : FAILURE in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/280cfae11f094c6b87ed67b482e3c67c : FAILURE in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c74b6cd9da6e474d84bc726ba8edace2 : FAILURE in 9m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6018aff358374191810581d15312555a : FAILURE in 11m 30s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/0bf5919a305047ad9a14542abdc3c9cd : FAILURE in 33m 30s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/08206c38d1424c619f7c5ea9721fdcf9 : FAILURE in 28m 56s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/5dc74e1a34e94937b8ada06ddf444c43 : FAILURE in 24m 46s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a7a62ce0a5234b0fa3e7b0ecb438edc4 : FAILURE in 19m 20s","accounts_in_message":[],"_revision_number":8},{"id":"e690cfedcdee01af80310db32d930f13de7c53f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 20:35:45.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"eda0bd927ee56f3683c8f98df617eab2efc0857f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 22:13:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4132c2762a644886b6e4107d508830ff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/70d64146cf07442ca28d85891819c263 : FAILURE in 3m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5dea6e2583a46399b41d95d44140e7a : SUCCESS in 1h 33m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc63e5b9b434440b968460a39252d952 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/388bcfb22eb049c4a5b0f2c6f851c8b9 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/390b18b0ae2849399690cb15a4abd9aa : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6900ce1ccf1343fb8c79f145cfd70565 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/269cc716c5e640eb8ae77fce03bb680e : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab684b4e7fe14572a12288be1c3c889e : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62e6db45a7fc4b4887a2dee1e145c839 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6ae5b34b3624002822565c1a8c3a785 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60f5e0281bf54b8c99fefd6e09aae614 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fc48a006a4c4546ad578636a875aac1 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d66fbb3447c4311bf4535ff9ae8cb71 : SUCCESS in 6m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae71305165da4d94aba0cdf4ee249ae7 : SUCCESS in 7m 41s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/b65be784d97d49d0822e303448698db2 : FAILURE in 24m 50s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d483944093d344128fdfa1f0874dea2b : FAILURE in 28m 24s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9c726986bb6d49638d959a01b22b0f02 : FAILURE in 49m 30s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b27721d92d0c48a495fcad798d99f979 : FAILURE in 30m 19s","accounts_in_message":[],"_revision_number":9},{"id":"b16aab190e717eba9b8fc9291646ef2de5fda21e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-05 17:42:22.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"9bb1d2d5b6dffa2db3f5f164f903fda6a8226301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 18:39:29.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/860a443f24ff4a51b5a7a10736a35c06\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/297b7057a07a4340971d6a99c96e3bda : FAILURE in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43d73d373cc940d09e79b48685517523 : FAILURE in 16m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c480aa78414917a8a365f250b653e9 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdaebd678f444a1a40a54438e8d1028 : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48a007685dfb4bc7bd13d33de5f9f9c5 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0de91f4463c74e3f93993c810a19db3a : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/263186e7554d43568f6875f9cf2eeedd : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03817bedf764462291709e46bccaa04b : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f39b22941f4404a98ce6bc512324159 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af80ac3f91b6464687350d61c1fc045e : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a11b351765f54d41827ccb9c05424cd8 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1abebbff59df4868a827d85681a54d05 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2beb5bdbf0b142fe96328795644cd5ab : SUCCESS in 9m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db01ce8768174e0f93dce77bdf6e8c50 : SUCCESS in 12m 32s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/347ebe66560f48b1af27d529f30de8f4 : SUCCESS in 29m 50s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/5dada3dd4ace47b394b1bcc6afcce3da : SUCCESS in 30m 22s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9e14fd6099dd4629a23b377755ed66e3 : SUCCESS in 31m 02s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a5cd04770de24ed099b1ef8e44934dc1 : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":10},{"id":"64136f905b2fc62007df71c09f5a0e0375372e26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-07 08:30:36.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"f78bf94ddb4b4a220ebfa468bd501c037c134b64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 10:21:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938782b12d134a91a2fb87971a002143\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b2a4c346087547a0bc7072d74f14a12c : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04e211bf32854490b5eac69f04e8973c : SUCCESS in 1h 47m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c11cbb842c14669b3b0fcd6bac67a2b : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57800693917448a08fff9192fb7e7644 : SUCCESS in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef5d53f0098e44e3986508ae403975ab : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a59fac61e654f8c9e68a142df9031d7 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd22c8c439aa415c8cdfccbfab9cadce : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb59a73c9c3f45cfb69f3cd858e2b95f : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44272bfc21034b85b46f76a68dd95b05 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18618afa664d45ffb4ef1778cc313bd6 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06dd5b4d34434988b38fd9877d0fdcc5 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8e34ee00c9c4290ac4993c22942d3f1 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b3f32247ed24786b12676c26635e1cc : SUCCESS in 7m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f486d0db6d0540d8aacb6ca23c4f59c4 : SUCCESS in 9m 17s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/f238f464b9ec4b05b04e047197613181 : SUCCESS in 28m 30s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/07e8f14ef7eb41acb6f7c24af00d4cc1 : SUCCESS in 33m 32s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/b949b1b304bd46999eecaef7f9588428 : SUCCESS in 47m 36s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0f6f27a9563643de9b902c2c9d6332c7 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":11},{"id":"32cc606a9d9640df3b2f3ded2523459210f51634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 15:04:16.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"fa0c38c5435aaf550b12bbcd6127d8f93dd865ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:56:13.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1557ee5d75334d3fa1b6415bf3ecc6bf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/073fc61be3464454ae3a8bf9603a86cd : SUCCESS in 3m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bd0511309f14d39b53e40ef409b38b7 : SUCCESS in 1h 42m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9079086757b04bc398144051e430ea1e : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c377dc0e9646cab4d00e64cf271d96 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b0025bc75d248059a398ff72688ea7e : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f7bfd6923894c3f93c1c79034899cce : FAILURE in 2m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36a05af9744042f7800397dfc8123e6a : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb67b994dfc84221b46ab4280aed815b : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d59a352675ab4f0c830b77d5352813d0 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dae312f78a64c5aa4492b263bf87a15 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1443a9528e645cca394f3c6e5b5af9c : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77ab8541c78e47c8845ee5f7497adf38 : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f447a5416514cd488ff6de7068b98f2 : SUCCESS in 6m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57f46648927845a096c85a43a7df34dd : SUCCESS in 8m 40s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/f658cc95dbbb469ca4bcd5f1bcaef9bb : SUCCESS in 22m 52s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/ee3c767b8e7b4af88a09ef3e3d9aec52 : SUCCESS in 16m 07s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/15908773e3d14ac3ac0ad73d1b9d9446 : SUCCESS in 32m 12s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/557015b8e27844cbabba7c45e9a199e4 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":12},{"id":"73283741c0259003ba5f741f2dcbedb0e5c59c07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 17:03:50.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"01c7d506dedc0404f7f3d2a6d65a15848e2a6282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 18:55:02.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/145071d884514129b731acf73ad41834\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a9909e0df31c4056bcc3e9940ee44f88 : SUCCESS in 3m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23f869a6e09f4548932271cbb332d26f : SUCCESS in 1h 41m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb901adbecf4787b6c8506d4eab994c : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a269fdff6ffc44d7b98e6403164a23bd : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bb5324a2fbb4642917d98635fe75d86 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/832cbed7414e4133a2e8beec9500b9b0 : FAILURE in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0acb4dd770c49bbab05c07ace061de2 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d4eec7c0c4c4c30afb08dfd91964714 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a714c1ff6f464d359c7bc5c87660af29 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0576184682134a138a6ad7dad0250872 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac05abc70722472fbd6f6a2145d8853f : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c338517582f4abf9fa49cd1238bc28f : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cd84446a0ce42efba17f13c0c03524b : SUCCESS in 7m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f21e21a3d0540bd9b944360a4868fdd : SUCCESS in 8m 36s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/eba325201e5e430cb7db5e88f4ac3144 : SUCCESS in 31m 02s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/f4c2d93c051c4fd494568907417621c9 : SUCCESS in 27m 44s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/86741bb0e48e44a4b5a126dacfbab80c : SUCCESS in 39m 51s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/14db3bb8d423490fa51096a24fc04216 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":13},{"id":"6bfdc22f9bb1a61938810f7dd203ab061e781bc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 10:37:26.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"f9692053db0908c672aae374c026fad4b3dd779d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 12:30:00.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d9e9c25261d4d3c8fd194d93b6f3382\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32cd39c9adc1447a99f00c2084893806 : SUCCESS in 2m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f91e5ea2c33a4b4bb380126ea7be71d3 : SUCCESS in 1h 33m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971892f4d997416ab8cc0a4ae7d61b3c : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93b43e54a464aa482d73aebe0a99f1a : SUCCESS in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9eaa387d21024207a2cd07094155f019 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e712f35532b498c99bed265313671a1 : FAILURE in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50ef39b9533841f5b7c3e77baf6020b5 : SUCCESS in 2m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c6897ae0889477ba0b7a532d403cccb : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56e0431f5e424c0e85ab3038ccd41932 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02550522d5a24b2993821ba3ac036a49 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfc23a8eaef24ed8afdc450d8d89e083 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb2c78e0c8f6416a8d5da5c32d1e0e35 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/924a279e14ef46bd8315cfe8a38e4885 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c281172c72e3417f903897df5373107c : SUCCESS in 7m 35s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/3dc4ea17443b423ab7a59c91736b4832 : SUCCESS in 36m 07s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/1eb00ec86888455e8d7a93bca5e95162 : SUCCESS in 23m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/19c6d2b665c7424bb84047e02ee23afe : SUCCESS in 31m 29s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/66d18af41e434be9b91bfb9663bb8393 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":14},{"id":"5b6179d60f9971542fdb775f462b935578e127ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 12:46:27.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"7ec6ddf4756614392fb679fe6824018f26d3ed54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:34:49.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581f67e8a8be4c9d83b74b5c0b1788e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0aa4eac379c546ec9a913b52c59056ab : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3c73efec7804f4e84510cbfae1f73e8 : SUCCESS in 1h 39m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e4ccda71084a35b868fed18522391a : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b42b6f5f4e34884987b48a09a9a8216 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e3ea274a5304a769ae0c931798179b9 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7847ee8b99f4c74a4b68d9522b436db : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd5972c73ce342fa96134175a3988826 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76e16323e5554b349ff7875eb21d32b0 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64d55ffc2f4d4247912cf005651119f4 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f9f99a1f18f4e648f49ad418dc3e8d2 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad01f51029e54cfb9baf3b178276477e : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11d35d0784844c1d968694b8e727ee3c : SUCCESS in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b9dca6414b04f27b8080aa9251ae8b3 : SUCCESS in 8m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0b837b180df4a5f86ba910c180ed43e : SUCCESS in 7m 54s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/c0dee04d0839404e8f579d02b598213e : SUCCESS in 19m 51s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6768111f0e364cc3b29d768afaa92e2d : SUCCESS in 23m 55s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6865949bd4914b538e5a2f733daed39f : SUCCESS in 28m 42s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a9d84b49e7f040778d005143de9387a0 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":15},{"id":"56b741db6e2b1e867834b13ce40895a06980c20e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-16 16:07:09.000000000","message":"Patch Set 15: Code-Review-1\n\n(4 comments)\n\nSee comments about deprecation version and switching to fictional project examples.","accounts_in_message":[],"_revision_number":15},{"id":"835b1f070b8ae33ab6ba5abcc590e8e01ee46c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 09:32:27.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"fa9aadbe70322e7c5417f437a564eb7de61b8f25","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 09:44:08.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"5ffb6d71f693867ee86ec031f2c6ddad938e4bbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 10:13:39.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"8003d815cd3b05802f4635f742cb03fbfadc19e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 11:51:30.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0b2ffc2c9f494e98e2d730c087ba4d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/86d3550ab7ac4d5d967434f39124ebc8 : SUCCESS in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e3067dd2c1848179b2a72b0032b9337 : SUCCESS in 1h 30m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ab2ebf115549db8cddf4601e296ba2 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba77f39ef57449e84b0ed5af8483edb : SUCCESS in 4m 18s\n- build-python-release https://zuul.opendev.org/t/openstack/build/19f4e9f5eca54501847de5b275212952 : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/321edbd07a5e4bc7b05ad2ac3be573ae : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13cfaf5c02e54b3488082e8817c71cd6 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5214e008058e40a088ee4cb0de413870 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a1b922d91a4427c83f25edc79322411 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ba54f8d7b2e47b09992cbbf3e16a2c0 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faa0724106244164bfe6a434ca409f90 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7290ccc20e1b42e18d84579e06b5ac6b : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eed3db9de6a543208ab7243915660208 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c7115833d0b4abba21b52cf99f5bf6c : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/964b0cce152a4b6a9c8ba21b5e3b9c9d : SUCCESS in 10m 21s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/bd8addd8a7e14ec19e4ad4dd2996e9ea : SUCCESS in 30m 02s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/fb0e0a2ecd96463e8fbf5eb3a94527da : SUCCESS in 14m 29s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/361ad387b2c64cd8a1e25eb1cd7786f5 : SUCCESS in 42m 52s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ba743467822e44d19de5ef112769c864 : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":17},{"id":"1ad6cd5e01007cf746a135a94d240f7b536ad322","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-17 17:12:48.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"886fa4b0eef8267f211a3276f8fa5fea83000dff","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-01 16:39:16.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a5ca1971f7369a3b01b0055976c191bbd7d6800f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 16:40:39.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"249eb5f526e54fae7101166f76620489a70e3120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:05:32.000000000","message":"Patch Set 17: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/949052 is needed.","accounts_in_message":[],"_revision_number":17},{"id":"496ea4741277ef508e47080cd05640be0da3a31d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-04 20:04:11.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"555e3bc6fc21f4e23094ef1772d1c7eb9cd432f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:55:49.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4c10ca441f42459bd0c141e0d888b7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc4fb04393ba4aebbbb2168a3bd14897 : SUCCESS in 3m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebe9af2d45f04fd993f2ad3611a6fe86 : SUCCESS in 1h 48m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21677bd3049c43d89e0cceda4e4ee632 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62dac0842c354b8abccdc5d2c57d34f7 : SUCCESS in 2m 24s\n- build-python-release https://zuul.opendev.org/t/openstack/build/35e86ca5bbda4861ba8768be12fe68ab : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af4128af8dfe41e895fce733e6d7172a : SUCCESS in 8m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a1d81b5f0d14933987b101207ffcd7c : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a52693e97a94f08810bce36e7406218 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd0f809f5067492a91925493fdf6b523 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0473766214ba4fcd900f1871a38dbced : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f37d96b1598a48cd995691071a55686b : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40d315a27f1a4838bb1cf4434840ae73 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/066c204be77449c7a44c25c788c2568a : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa41064a6e994ae09dbcfc8e9fe739ea : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfd9fe231473463ab0b27a065f636296 : SUCCESS in 8m 28s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/561ffb0076884b8f82ec47781d8c35db : SUCCESS in 26m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/29e48a25696b4fdc9bb07426108ce23a : SUCCESS in 32m 36s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4c56c9007b724df984f3dff489057698 : SUCCESS in 38m 09s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/327f519442ac4c3d8f146ea1f27a47cb : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":17},{"id":"d50a65aa8bfc07c37286c5d7dbe710f8227cb63d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:57:05.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"2c024b1e79ee6b3c7b47f6d9546cf3e88057ac52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:41:24.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea94d347d0346e6a8c50133d0643182\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e718b172dc444fab819b53f20b673c70 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21ef63e2352f4570bae4bc22a9cd1786 : SUCCESS in 1h 37m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e638d5528be541f5a0facaaa18aa1486 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7621d8817d4476094c09fe4706d3d09 : SUCCESS in 4m 03s\n- build-python-release https://zuul.opendev.org/t/openstack/build/48709b71ac4144c5a105bf09c29a937e : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54924a16e278445580313e919aa7b8f8 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19f80f11c40a4bc6a020586a670f7ad6 : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7a22aafaadc4e9aa87037df588a8441 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70bb38e1e9594d4daf528f6555123495 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c94b8ac7ce6342adaf37013eb3cd431e : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66e60014960542669a1b11bb473bb06c : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a0b5c38bf88489891e5336985575eed : SUCCESS in 10m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fe76fd381ab4482ad5bcaefe4dccf20 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ef450b6736742f7b6c71046efab342a : SUCCESS in 8m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f23c848d4a94a0b9bd7a6e6d8e10dae : SUCCESS in 8m 23s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/1a2f29cad96f4d949f8cc74c4b1924f8 : SUCCESS in 15m 56s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/3107c85a9f774b84849994bf8534e2ea : SUCCESS in 34m 55s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e90a54f29c674dba816ed12449e0c232 : SUCCESS in 50m 07s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/21c2912076594b9da08cfe12ce40aab7 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":17},{"id":"426ed9c922bab1dbfd4ac438d8715f856fc0e8d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:41:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"d4e1eb1120cafee30115f0c5e0d9c7e9b90666b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:42:07.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab78b5245ff480faffe180bf2c428a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a611f695f9714f35b398fa1049aac99a : SUCCESS in 34s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"370e3804ab13999db72c26aafc6167e2c68f3bc9","revisions":{"6bd9d296a03e9694fa531eb6d41a72b3da5430a0":{"kind":"REWORK","_number":1,"created":"2025-05-07 16:30:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/1"}}},"commit":{"parents":[{"commit":"d15543ef27b0136a1c692113dafaad89daf9861a","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d15543ef27b0136a1c692113dafaad89daf9861a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:28:25.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6bd9d296a03e9694fa531eb6d41a72b3da5430a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6bd9d296a03e9694fa531eb6d41a72b3da5430a0"}]},"branch":"refs/heads/master"},"7f8dfd6e93ab75042e9483d9c05f183d20e78fdc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-08 13:29:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/2"}}},"commit":{"parents":[{"commit":"3a359de995a58c6ddc2579afd68fe8c45f3d9983","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3a359de995a58c6ddc2579afd68fe8c45f3d9983"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:28:59.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7f8dfd6e93ab75042e9483d9c05f183d20e78fdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7f8dfd6e93ab75042e9483d9c05f183d20e78fdc"}]},"branch":"refs/heads/master"},"6d29bdd7c7242c6787daa835e20d4992ab17107c":{"kind":"REWORK","_number":3,"created":"2025-07-03 10:30:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/3"}}},"commit":{"parents":[{"commit":"54a60e6159e43c5cca8d2c5d8d482eb99a82fba5","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/54a60e6159e43c5cca8d2c5d8d482eb99a82fba5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 10:30:16.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6d29bdd7c7242c6787daa835e20d4992ab17107c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6d29bdd7c7242c6787daa835e20d4992ab17107c"}]},"branch":"refs/heads/master"},"c3f9244082ef9f24e1a5b56d3122b899e2cf4bc1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-03 14:37:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/4"}}},"commit":{"parents":[{"commit":"0cd59ab634aa7a120ddda3ea9d8c8b2ca0e41ac2","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0cd59ab634aa7a120ddda3ea9d8c8b2ca0e41ac2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:37:45.000000000","tz":0},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c3f9244082ef9f24e1a5b56d3122b899e2cf4bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c3f9244082ef9f24e1a5b56d3122b899e2cf4bc1"}]},"branch":"refs/heads/master","description":"Rebase"},"5587ce76741fa7083a085c6ea6e8efc80cf1b483":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-03 14:40:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/5"}}},"commit":{"parents":[{"commit":"ae8cd3f90baa2266c8ac00b66a9d6d741773dbc2","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ae8cd3f90baa2266c8ac00b66a9d6d741773dbc2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:39:55.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5587ce76741fa7083a085c6ea6e8efc80cf1b483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5587ce76741fa7083a085c6ea6e8efc80cf1b483"}]},"branch":"refs/heads/master"},"55bdbe70101a80fea6e227cf03f8fd3df568d885":{"kind":"REWORK","_number":6,"created":"2025-07-04 09:13:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/6"}}},"commit":{"parents":[{"commit":"8826cc3cebeea553665ad15f84621962c4622d95","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8826cc3cebeea553665ad15f84621962c4622d95"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 09:12:02.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/55bdbe70101a80fea6e227cf03f8fd3df568d885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/55bdbe70101a80fea6e227cf03f8fd3df568d885"}]},"branch":"refs/heads/master"},"1ff8e20ff7db7cb14565cb23cecc14c07502b965":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-04 11:00:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/7"}}},"commit":{"parents":[{"commit":"45df76c3e61145c107c6aef27628da81c014d9d0","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/45df76c3e61145c107c6aef27628da81c014d9d0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 11:00:30.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1ff8e20ff7db7cb14565cb23cecc14c07502b965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1ff8e20ff7db7cb14565cb23cecc14c07502b965"}]},"branch":"refs/heads/master"},"04e1dee12308cf040b1b5028add5d148bc64c5d2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-07-04 17:59:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/8"}}},"commit":{"parents":[{"commit":"51d50d477b9add8ac644707c27b3a1322c71c5a9","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/51d50d477b9add8ac644707c27b3a1322c71c5a9"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 17:59:02.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/04e1dee12308cf040b1b5028add5d148bc64c5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/04e1dee12308cf040b1b5028add5d148bc64c5d2"}]},"branch":"refs/heads/master"},"c044cc3be80b5c45f872c75042f908cb3ebac613":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-07-04 20:35:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/9"}}},"commit":{"parents":[{"commit":"8468c9c6a05564a29ad744817cf8feb3b1565427","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8468c9c6a05564a29ad744817cf8feb3b1565427"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 20:28:47.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c044cc3be80b5c45f872c75042f908cb3ebac613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c044cc3be80b5c45f872c75042f908cb3ebac613"}]},"branch":"refs/heads/master"},"a1a60102c4e50760312e9f1ba8d9704873c875f0":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-07-05 17:42:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/10"}}},"commit":{"parents":[{"commit":"d92b9a5af334c3430cc74634d81ab9f77c3d39af","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d92b9a5af334c3430cc74634d81ab9f77c3d39af"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-05 17:42:04.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a1a60102c4e50760312e9f1ba8d9704873c875f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a1a60102c4e50760312e9f1ba8d9704873c875f0"}]},"branch":"refs/heads/master"},"7f7f28207e76ebc676db662d2fd2b57d30d787a7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-07-07 08:30:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/11"}}},"commit":{"parents":[{"commit":"6ee09af502e1dce2a82e6204e4e1160bce574542","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6ee09af502e1dce2a82e6204e4e1160bce574542"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-07 08:30:29.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7f7f28207e76ebc676db662d2fd2b57d30d787a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7f7f28207e76ebc676db662d2fd2b57d30d787a7"}]},"branch":"refs/heads/master"},"838abb0181b4e2a4b4bb22c1f70b64fd173a956d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-07-09 15:04:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/12"}}},"commit":{"parents":[{"commit":"42206265e2195be0059eca6ec7737f2c71b68297","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/42206265e2195be0059eca6ec7737f2c71b68297"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 14:27:31.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/838abb0181b4e2a4b4bb22c1f70b64fd173a956d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/838abb0181b4e2a4b4bb22c1f70b64fd173a956d"}]},"branch":"refs/heads/master"},"17b9aaf1327b63f5b7fce41f2b6f6dc612c6686f":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-07-09 17:03:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/13"}}},"commit":{"parents":[{"commit":"124d90f5a5e24b301b87501b9940dd521bf1d31c","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/124d90f5a5e24b301b87501b9940dd521bf1d31c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 17:03:38.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/17b9aaf1327b63f5b7fce41f2b6f6dc612c6686f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/17b9aaf1327b63f5b7fce41f2b6f6dc612c6686f"}]},"branch":"refs/heads/master"},"8870d6f4a442ffd041f20082e8aca544d0d49417":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-07-10 10:37:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/14"}}},"commit":{"parents":[{"commit":"47763d404b4b3aece2932bcac0379f4230301de9","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/47763d404b4b3aece2932bcac0379f4230301de9"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 10:35:29.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8870d6f4a442ffd041f20082e8aca544d0d49417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8870d6f4a442ffd041f20082e8aca544d0d49417"}]},"branch":"refs/heads/master"},"177205a25ebcdf9150079efdcd6292b8a4516423":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-07-10 12:46:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/15"}}},"commit":{"parents":[{"commit":"d199692a19946712311c3716bc29437781003db9","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d199692a19946712311c3716bc29437781003db9"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 12:41:23.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/177205a25ebcdf9150079efdcd6292b8a4516423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/177205a25ebcdf9150079efdcd6292b8a4516423"}]},"branch":"refs/heads/master"},"761a9af693f7f35fbf7a5ff5d240879a76152c6e":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-07-17 09:32:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/16"}}},"commit":{"parents":[{"commit":"b22e69046aeb76395822c977252db2711f244ede","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b22e69046aeb76395822c977252db2711f244ede"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 09:32:05.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/761a9af693f7f35fbf7a5ff5d240879a76152c6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/761a9af693f7f35fbf7a5ff5d240879a76152c6e"}]},"branch":"refs/heads/master"},"370e3804ab13999db72c26aafc6167e2c68f3bc9":{"kind":"REWORK","_number":17,"created":"2025-07-17 10:13:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/949053/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/949053/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/949053/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/949053/17"}}},"commit":{"parents":[{"commit":"9e7b9cc268320f937ceec48dc0577ce39634d02d","subject":"Deprecate \u0027[entry_points]\u0027 section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9e7b9cc268320f937ceec48dc0577ce39634d02d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 15:37:50.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 09:54:49.000000000","tz":60},"subject":"Deprecate \u0027[files]\u0027 section","message":"Deprecate \u0027[files]\u0027 section\n\nChange-Id: I96e4cc5f7697bbae398bf496b0111781a1d1708c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/370e3804ab13999db72c26aafc6167e2c68f3bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/370e3804ab13999db72c26aafc6167e2c68f3bc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
