)]}'
{"id":"openstack%2Fpbr~949055","triplet_id":"openstack%2Fpbr~master~Icefe8027567cf7cc8153552712931d68174dd99b","project":"openstack/pbr","branch":"master","topic":"pkg_resources-removal","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-04 23:41:30.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-17 17:14:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Icefe8027567cf7cc8153552712931d68174dd99b","subject":"Deprecate various \u0027[backwards_compat]\u0027 options","status":"MERGED","created":"2025-05-07 16:30:11.000000000","updated":"2025-08-04 23:42:47.000000000","submitted":"2025-08-04 23:41:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949055-pkg_resources-removal","meta_rev_id":"2216998cb458ab87d0ac17db6d8c829e5bcc73c0","_number":949055,"virtual_id_number":949055,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 23:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-08-01 16:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-17 17:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-08-01 17:00:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 18:08:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-16 16:12:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-08-01 16:43:59.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-08-04 20:04:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"62082a3ebadb3d5a5f87293a90f0cb4c516710e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-07 16:30:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e305e80646489e40328ba396a5fc84870658dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 18:08:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c758d44434d41b9bb92e6fc257168f9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49d722969c5a43539fdf2d9ef6e33293 : SUCCESS in 1h 33m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac4b4ef96fe44d39a0313849725087d : FAILURE in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29001b52e984403be774cc009682922 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5404ee7c2a44c8e96f53b27e24039ab : FAILURE in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a511823fde1943f59ff23fe691525405 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c690d7ba603149a18ce3ad903c747789 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bde4465b64045b0a45b8eb5c8b5f94a : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e078d968204f4e40bdaf61640c89c6ee : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc6e12eb27124039abe96e8baad4e878 : FAILURE in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bb7da93cb0b44da8d90234c027cd4c6 : FAILURE in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37748c50b1ac429fa020835502dffbac : FAILURE in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d99772e4a68b4260b86519716814c4b9 : FAILURE in 5m 02s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/03a158cb88e14c3db83e555d70dad36d : FAILURE in 28m 04s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2b25fd5c2bda4b54be59d6970519fbbd : FAILURE in 13m 30s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/8622ec372f1346dcb2d770ec5df8824b : FAILURE in 15m 37s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/764cda3517db457faa1057ff0a845769 : FAILURE in 25m 15s","accounts_in_message":[],"_revision_number":1},{"id":"839c9d808e6901ea5f95161957efc52754d2cf1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-08 13:29:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2fb7b2488223c42fda276dae0ebe5586f2c45b9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:08:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f60416d089b458aa400e8636d0f658f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8893a7845d1147bba47fce9051d18697 : SUCCESS in 1h 35m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8407b31294d4760b752bf2bb3283f66 : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448aa1b0373447149ab896956a802beb : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a738fd0bb2a4115a7905363db8f7fb2 : FAILURE in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3153767fcde54d61ab9c5cb8e39b3e3a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36a7cfe0e8bc42a8a9d627f373224e45 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d72fbf5cf824d929ead26ff219716e4 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8858d23d3e747d1b65b34a305ef44b3 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db5eb1947ce847b989d14980ca29a8a4 : FAILURE in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98255812f8974dd0bc2a01cbd0823cc4 : FAILURE in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a14e4c0572454946bd74778b831aa02e : FAILURE in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a02b941f7784bbd8f1dcc9829daea86 : FAILURE in 5m 32s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/b35e59b683b24fb39e046bfb8819cd42 : FAILURE in 15m 06s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/7af810e8258c479c8507cfa5d0693ac6 : FAILURE in 24m 29s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/fd895d504ee3410eb558a2306f98189b : FAILURE in 19m 45s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/f9e419a5b4af43db80082db538633c7e : FAILURE in 21m 12s","accounts_in_message":[],"_revision_number":2},{"id":"f824c935f2b76f5ab7ee467c7f64809f5b56bd9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 10:30:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4b2d641ae937f374824658ae962257262765913d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 12:33:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cea283bbbd247348eb37491913f606e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a121c0a73024275b7df8f53a4a1a22e : SUCCESS in 1h 51m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef9cbefad824596bfa54a554554fe4d : FAILURE in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da8374c84ace49778ec0017e9f8cbacc : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d655850e2e242ba9e27422463348f51 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a877b9a5553488d8efc5050cedca9cb : FAILURE in 2m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bb0531e609147248e51d05a93e72050 : FAILURE in 1m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/486aea778fc64495b07144db55eae16a : FAILURE in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1717b9a1e7644f4e9be38a5fd70b8a19 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef99e3cf68f943d8ab340859c806e902 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b87f06018b164ed0b3403a637f6f5eb9 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2724bc8eeb3481bb5112ecb526b8b4e : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1556e9ba24a343bcb8eda628e77cae7d : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14b9cee9a7404f619bce8cb01fd7acea : SUCCESS in 7m 39s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/c739973eadf84194887a4bf88aaa7dff : SUCCESS in 31m 13s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/1e7969b80ee6479484dba7c1cf270477 : SUCCESS in 22m 45s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9f5f24379b9645e48c766b5be4d053b9 : SUCCESS in 41m 24s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/1d3a16cffef94a998fa1d2de3f21d4f5 : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":3},{"id":"befb867d0bfd3bd274020a3cd76862f4b0be31ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:37:54.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bc4b5db59299f460eb5f49597f915bb73b422ddd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:40:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"45b4ee911f016f87811c7e061443febcc0711440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:43:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7d3ad3a74e344bdbce20abb1f436f6d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5a9bfe301da4c9587acbd5a4b043e5f : SUCCESS in 59m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff298bc1fc0c4de280a6fbef7eca76a5 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ab2264713045978a259d9795232605 : FAILURE in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3796a5a00e7a41c1bd5c892641f263f9 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f706e656c6474d3f87dfd49c5fb75511 : FAILURE in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed7db1ca1bd04e7195fb1b6a0302cb89 : FAILURE in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e463a5d48e14cdb8d7b922b85139134 : FAILURE in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d26653edbc3e44458427fde679e9ecb3 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71375ef6f6884e41b053f60f8d15905e : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12b867eb7e914f3499d47d768a5a6ba6 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80eab90aa6374b019a3049be8e33d688 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1276da52b9b242e48b4923517f756b73 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75cece5991384aa7b44fafbbd5a4eab7 : SUCCESS in 5m 11s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e6705805e0e84cd4abc321f1c710baea : FAILURE in 7m 57s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/cef9e8e021bd4ce387392cea1134c960 : FAILURE in 7m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7aad5cd0f650430aa8b41173fa8d61a5 : FAILURE in 15m 12s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/8d850b99b76f49abb965d7dd0ecc9abe : FAILURE in 15m 15s","accounts_in_message":[],"_revision_number":5},{"id":"6c2f7393a871a41e2cf9ef920f968bf57f3b9ffe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:13:15.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1c433ad6ba3c5d4f567e29d24bf4d488fc1b4e52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:19:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1c79041e4ba4505bb6740d6d2aba8da\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/126a2b0a897c4330b8b1c85bc468c2b9 : SUCCESS in 1h 01m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ee8c3f364564cf6bee35ad6c7ec0aee : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f5d4a8de744fa9a7e9337155864ecf : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dea7d35206745cebbcc1dd00d813e11 : FAILURE in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/311ff0ed2dba4fbfae17ccecb51da520 : FAILURE in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4735375ccbf04b9384c0a6aa7d547811 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a959991eb6044057b07cff80cfa993bc : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d83e5f5b7db8477fbb339a6a5bb8957c : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d610619fe8ed49e9a6add7bc45831b33 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04826dbca9384d3ab84ffb4a516b0858 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c272713e68b84dad8a50adeb8d38c702 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d46c34099e0f420cbe8cc30ddbbd2cca : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a14732e93c53480a9db8fe7a9704f697 : SUCCESS in 4m 33s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/074d3a6c7f414c169641892cdd4f186a : FAILURE in 22m 00s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/ccdf4e12a80c41e78e89cd81e6225a95 : FAILURE in 27m 03s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/09615e6e019c425681044548b5f83d74 : FAILURE in 33m 01s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/8efab045565b4fcc8e12bce1a5f7c071 : FAILURE in 40m 51s","accounts_in_message":[],"_revision_number":6},{"id":"ccd91d98594fa810552c36538295e0730e13fbbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 11:00:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d6eb9a5804ae6f01dae2dbfaca67533c414850d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 13:00:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b6d3b7f9878480f8f9d920204557018\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16a8a7bedaf144c5930c66dec936bb98 : SUCCESS in 1h 39m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fe10e67e984cd3ad3192497c6d0332 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6153d16a52ce4b0180d4649601ec4ae2 : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28b18c4aea934955a2be35f2978aa26d : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d2698de983642efad2b3fa10ec5677a : FAILURE in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50173e40f8794647a83f318815f90313 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97c778b44b944ba7936a55f2b614574a : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c76144d840bd498684311da9c991a025 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c9e04bb64cf4811917a34fd29a071d1 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8988e070cbaf491797d0a25312ff396a : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2cd022ae22b498aa14b789419fbd31f : SUCCESS in 5m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/699401ce1d834e07917b16843e2c1364 : SUCCESS in 8m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85ac86bca37d4da7a7a97c615a119b61 : SUCCESS in 7m 54s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/579358b0bc3944b68b71a8417722d10a : FAILURE in 21m 11s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/8177b07bc6414c04bc55c17dd5f86796 : FAILURE in 24m 27s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/b1d96ac6f1184dc294168b0da95c5bc8 : FAILURE in 32m 46s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/3160b1eb1b3c4a6dbba9884e76fef9fc : FAILURE in 43m 50s","accounts_in_message":[],"_revision_number":7},{"id":"5b2a7c03ed5b3cdd662fa8fae75cb0854f2947a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 17:59:08.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"692af309c87b200ff99f45ade7b81019f81bf925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 20:00:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/388389aca02547c1bd10fc1ff110d8d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71f1b4aeb7b842a78885f06e69ff9d51 : SUCCESS in 1h 53m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ec5a2a9aac4d79bdcd869195d97266 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0946ac33d5de49f2a1e7adc03d88574b : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d294804cc01e436abab4cb668c80082d : FAILURE in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e370ee514c2404aa4a666cb9222e074 : FAILURE in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bd1252ab15349438c9968ad2d889792 : FAILURE in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/293a3915ee564ee8b649eb3b10da7f47 : FAILURE in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89c081c92c2d430bb1bad4b1f951426e : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b43a956996e4de584c6d454b5bef515 : FAILURE in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a248acd19db24ccca426460b1125c992 : FAILURE in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67ff4f59fc044d34927ec25968401f6d : FAILURE in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd1e67de2c5c4aa581a405523ffc733b : FAILURE in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9fef58bb5a249c18129f8b7f63b84f8 : FAILURE in 7m 11s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/cfa4c49bed2e410aae111f3e2a182218 : FAILURE in 24m 45s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d58c835f367042d3bd90ec16d4d964d1 : FAILURE in 18m 13s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/5b4bb1d16e0242388fd4f621ea567ef5 : FAILURE in 40m 12s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b213a391a6034e439428ca7000f77484 : FAILURE in 30m 56s","accounts_in_message":[],"_revision_number":8},{"id":"fd427bf7a2bb0784dd24fc7f8eed5b2bb5e078a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 20:35:45.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4e63967c4f3dc477a450bd96d24b0d2969a8f404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 22:25:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e940c1908d54d6c88ea9a34e648fcd4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22e0a2b2606647838176c6493ffd376d : SUCCESS in 1h 42m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77fe933b7e8c4ebe9064b285c6234ebe : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4404522dd0ec4ddc94d0e13421e8b080 : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fae41e3f86ab43599552b44aab4453da : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a10b0ca81c644acf8836ddaac4d7bbe2 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0484738542c7435cb7158e98a321f256 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53681fed764c4ad48ed64e07ed2c6b35 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78fb27848cdf4e53afe9471e3ca032a3 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6987b846dc0245d0b559a8fc88380863 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ba24944cd214162951408c557bb238d : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e902cb3001c949a08c42f77f4d3d364b : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f850d8b832f4427fa131e632c9d95d08 : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c376f87532a4163be645ea4039e6ea3 : SUCCESS in 7m 58s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/c1c50d9b94f642b89d483f2811c07f5d : FAILURE in 27m 22s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/fa34cb41417646f79d28090b57b51d96 : FAILURE in 25m 40s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/424ac071da1643c695a729a6a43dbb4e : FAILURE in 23m 20s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/e5ade23e3a974e86b1779ee0d9c684c8 : FAILURE in 18m 13s","accounts_in_message":[],"_revision_number":9},{"id":"2a8d087108e59e1d9881c597104becf493b4f8cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-05 17:42:22.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"dd1c928d25e2552c9220aefe96103d74287daafb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 18:59:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4f81ab8c4e54786a14ab4e13ab3449d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2181e2b7569d42b3b17f10ab984ccf72 : SUCCESS in 56m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec1a3e905db4c8eac7472de2ffec063 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bff3130e16e43ea8eea13034d80d5b9 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8643218b75d347e082165f9461c8b362 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf4dc0e4cf71493c8cb6bfbb7b32c881 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efd32ec392834401bae7e2552e7519c9 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78405bf8cb8f445a93a13a04a739a175 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2faf3680a7e4aca92cf03175dde46ee : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc8d0ed7a87d407a8e4b1b33bf734321 : SUCCESS in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d04e8f4fd2c44515aee41ca837af992d : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6fe76334a7846f4aa24b6e6f02036a0 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88c8a70af7d94121b72da3a17ba2ec0b : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/569bd92c60e8426aa90ef2d91d5740e8 : SUCCESS in 11m 05s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/35993498c20e4f459db1b77d01f548f6 : SUCCESS in 27m 29s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/7b91ea2d46ee48aaadfa08d7e0be9ee3 : SUCCESS in 19m 53s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/13925e165d2f468a961285bf55b1025b : SUCCESS in 35m 53s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/6d1aec355e5c4b7cbb196c5a5f6dcda5 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":10},{"id":"0046e6ef4774b87717f0a2178629a56d358ad1c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-07 08:30:36.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"64fdc02ed28c1ad976db0a5c5366b779ff63a0ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 10:10:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e821821d04244e2e848876d4bd778a59\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44143580cf52447299a20de10d11ccae : SUCCESS in 1h 36m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38b3854715a84a1b9a1ac21d9df4affd : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4429547fb11420b8a1c4408adb47160 : SUCCESS in 1m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d6d3d2f0fc4441957d42f497f5fc2b : SUCCESS in 8m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/876a26361e884649ad7933be08ee665a : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd012fce571847f8a79e3f1cf3ff2be3 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbe86a53fc3b456fa9bbaebf1a2664d8 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90160e319e58424ca9e8426af81eae7c : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81d5b29084664ae7804d6d8bc3dcb3cd : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04aab859f3144a9eb7fb9e62edf02d33 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb924562ad17432889bf731661568844 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95c8d12e2bad46a1bff69ab12cf740f7 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f3f59d33a984604bf5332e3503b1889 : SUCCESS in 4m 15s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/d01a2937333040d4ad903781e50ac9b9 : SUCCESS in 23m 54s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/95785946012d46059d3ccbc1fdabd4b2 : SUCCESS in 16m 31s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e879debb67a847789f4a5a7671fcfad7 : SUCCESS in 41m 07s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/6cd9418f73074838bc0a01fe8b43e0e5 : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":11},{"id":"5c7e26035f8f6812878e06a94170b5c64ef2b9bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 15:04:16.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"4be061e050f78cddb72736bee94b0588275cad50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 17:03:50.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"0ba37cbee3ccf4d4b9ed643f38a3bb1d4ba0d215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 19:07:23.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06af1736af1d44f8b5025b25be564165\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57e8ba38aed34ed08d5052b3a998a65a : SUCCESS in 1h 47m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd654819a64849088ef954e167dc0266 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb86eba44e84ee793d0dae529682f0c : SUCCESS in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88dfea3c801a48a58e841563e2507e75 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2390f58dc03c4ecbb6c862f7b7e0075d : FAILURE in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21003862850b4eef8f695e6c371d61da : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf3d8f7fbbdd41ce920f2d6e9d1a0c4b : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d4f5a5b23c64a69acf917c6b17c9f26 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c67e1dab4df476a9ee6ca80eaeca951 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b885e853943b4f9b865cc7fff7098d01 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78166ff79b404a3b836cd009bdf04505 : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ca6b5b133484684bbd45248fe793d63 : SUCCESS in 7m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07106e2ace764f8ebb9cf196f3cd45c1 : SUCCESS in 10m 23s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/368b99495b0c4630b06be96bdab00902 : SUCCESS in 22m 48s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d8c93f1c58d249a0aa0a3f576b966778 : SUCCESS in 20m 04s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e60bd0c31b07441d813d8931727a680d : SUCCESS in 27m 43s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/18e13cb516f94d5f84c7fe79ff69939b : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":13},{"id":"542099be77fdd9244b38af1960de198a943677e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 10:37:26.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"f1c9a76da54b129548825e5761867343c10d369e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 12:46:27.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"7adcb01d59a787e62c5f744ffa8142ea011a548e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:57:59.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2eaaf1cc5f44a1815f761746166965\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edd9c1d85c794c4bbd5fb413a51f0d1f : SUCCESS in 1h 37m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e28ebb6b16a4b93a691daf6ce1d7477 : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf752299f54541a2a6d847017d993873 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/839c556c8b8f4e7fb1013ae6cd08e4b9 : SUCCESS in 8m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88775dfc23a049a294c42f772b710943 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c09a4875495b47c5b7de18e2bf07adc9 : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/995a64fbc5764f7cb9061df14a7496be : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7147551107c4a72995c1ed1f761cdd8 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b15ea83e9acb468fbfca8c7c6535d0bc : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e47d2b44f11547e298eaa6ce3375a2f8 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fcf4759391264465822a2bf38a10559f : SUCCESS in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc072d20228841c29c433e43041f2c6f : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dc3f632d13e43b1a919dce056e340ad : SUCCESS in 9m 53s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/eca414c3c5104a8999c3360eb4a1d845 : SUCCESS in 16m 17s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/959ae9f8345b4ec8a3e468e116b5192a : SUCCESS in 13m 48s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6f0d59a93b744b58925c0566344b53f3 : SUCCESS in 29m 41s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/d16aace4b2a84a96aa4fc6d6cdb23765 : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":15},{"id":"ef4a14f56c3b36b3d988fb8d0874972572db162d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-16 16:12:49.000000000","message":"Patch Set 15: Code-Review+2\n\nDo this and the parent need any kind of deprecation documentation or release note?","accounts_in_message":[],"_revision_number":15},{"id":"2ca89c86af61d353ef3d33d88633402fe993e082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 09:32:27.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"16e5af87ad74b356b10e13245823153a646e4925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 10:13:39.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"4e18aaf80ab0a35db88552db13cce8b8dcd1e707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 11:40:55.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f8c0daf9e04ee18e6009576f72dedb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6aab405fb3b4134b706edc4b350b1f7 : SUCCESS in 52m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3968ad4e946a4f10a41d76a91ee7cda0 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf700dc258446c085f65130f7757fa9 : SUCCESS in 2m 16s\n- build-python-release https://zuul.opendev.org/t/openstack/build/c2a6708f84e14762afe5d61897717b83 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/743ca1a47dae4f61a1de2c41085f7a8c : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/207701a7d11141d7bd8752720441b931 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41c68039b493408c819dda92162f3b2c : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e80ebfbd7d6245f99182bcc8eac5ac51 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e51fa7b2ed74f5493da4a2eeda9f46d : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fb36e2a41ac4a4abb217d61c0df8494 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20ce909b49764704bd1419861a6e8fa2 : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72b804c2e8b24417a5f5d138612ccd4e : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ab56d7bb2c64b54a7c783a3a3b10cb8 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4efc0e78f1a04f6aa6d174984f8c4993 : SUCCESS in 12m 12s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/63513cfe4b1f4b1084709a06c4ca1ff2 : SUCCESS in 21m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/3b6a457c539c40489980149cfc5c177d : SUCCESS in 24m 13s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/64e48e0c3c1f4daeb376e01ebd3db8fe : SUCCESS in 29m 40s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/af7db62977c14ebe89550dff6f8d9202 : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":17},{"id":"1d2d327a801235b32707eefa712c562072bb2da3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-17 17:14:33.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"b32da32ca3c62731d438cce946931eac4bbbe0d6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-01 16:43:59.000000000","message":"Patch Set 17: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"edeae9a72bd62facb182ae75896983bd75e1b7a6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-01 17:00:00.000000000","message":"Patch Set 17: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"2f54752639fae7861770ca08dc25367bf522da7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:05:33.000000000","message":"Patch Set 17: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/949054 is needed.","accounts_in_message":[],"_revision_number":17},{"id":"5900385460f11f28e7f2c9d67860f70e03875d5e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-04 20:04:49.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"873fa7bd845e08dd612c3e811655a2e24fca4906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:47:31.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f68ae510defb45a19a5771dd1745ddb6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/555973c7e447479ab82949439f5cb152 : SUCCESS in 1h 39m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef7f72d947943d9871afc9504b93832 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20118087a73c4523b96d6710af50f213 : SUCCESS in 2m 21s\n- build-python-release https://zuul.opendev.org/t/openstack/build/a56fd40eff2b4c39b8cbe72c27624fb3 : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b906960c08148e8af1ad7942625f9e4 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2629258f59d3435e909a971461172dba : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/259ae01e992b4af7a4d3031b584ad50d : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba569863691a468493f0859c6f66f5b5 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27760c18535c446c9eefef6fcfbeed46 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4a2e74de19c4a089e0eda3ae7f5758f : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40a56f24ca9442e18b0098c5475dac39 : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae89243d7a9f4affaa90333f14c3ed26 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b64423fddab64c12ac6a70c4ac9a4b38 : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/229a6c26d0bd477d8245bf2e2de138f4 : SUCCESS in 10m 56s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/09f84000b6b144a3ab70ff58ec9298ff : SUCCESS in 28m 28s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b420718a98914887a8bdb33b15308e72 : SUCCESS in 32m 54s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/11e19e1e971b49ccad98734aa616bccc : SUCCESS in 42m 25s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/494100d042b64d2d93da85f42a5b410f : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":17},{"id":"2ea4ec8e4101e0ce30568e3deef177cb40d2a1c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:57:12.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"3e864c2a20a9abc43ee9fe22146128f06f99f3eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:41:29.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ed96b32f75a448d8ae273bcca12c94b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21809f004eaa401c8aef2258888ddbf5 : SUCCESS in 57m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492a56c9f1c14ec3b7f2132fa5cfce1e : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a23013526c549fd8a0364273dbf0f80 : SUCCESS in 2m 17s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e4d98d20431f42aaac274b26bce88c5e : SUCCESS in 1m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/323ffc9e0ba54264ad394646f7aecf08 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b9d9e1e98224241a249986fce3947c4 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d12d8904db4d65a6809522c23e5e24 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d48440fbc0ba4757851c53f3d88a1412 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f20ef502f7456bab504c32fd14a37c : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a91ff22ff0544181ad419ffea7843359 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4b86ae2699148e08df330b2ad9a7c85 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ea4c0cee9fc46b982742bd5be0e0ed9 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0110fa57c1b4202a86fb86e9f757cb5 : SUCCESS in 6m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d145316413764dffb3d5a7d16de47ac5 : SUCCESS in 9m 33s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ceb942a303af46a68190a5cc29fb4bca : SUCCESS in 29m 15s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/88f3555245d64fe9bc555d16dedd633e : SUCCESS in 25m 27s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/b3ceac12675e48efbc24a3d102ab8a1b : SUCCESS in 40m 12s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/68b812c7db8e47a497d5d03ab4624014 : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":17},{"id":"5e82e7bebce1a5a518d745295d38f7c0e1903a65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:41:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"2216998cb458ab87d0ac17db6d8c829e5bcc73c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:42:47.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f8aef61412e4fdaabeeb3a0f3f26a87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17b27699b1348f88c740dd0ef95f10c : SUCCESS in 33s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"8f1a6ec4254d005518b47edc62e2916c4e3d2a57","revisions":{"2c9076d209b3bbf9442093d22f00b88bf8f9f93d":{"kind":"REWORK","_number":1,"created":"2025-05-07 16:30:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/1"}}},"commit":{"parents":[{"commit":"24cbed9f74f3a82be4482860083bdc34dbc6429e","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/24cbed9f74f3a82be4482860083bdc34dbc6429e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:28:25.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2c9076d209b3bbf9442093d22f00b88bf8f9f93d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2c9076d209b3bbf9442093d22f00b88bf8f9f93d"}]},"branch":"refs/heads/master"},"5a1b342d7074ebf4923bba0765a8f7c32c06763b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-08 13:29:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/2"}}},"commit":{"parents":[{"commit":"847827fb886e1e61686e4c4e5591f47bd667dfda","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/847827fb886e1e61686e4c4e5591f47bd667dfda"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:29:26.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5a1b342d7074ebf4923bba0765a8f7c32c06763b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5a1b342d7074ebf4923bba0765a8f7c32c06763b"}]},"branch":"refs/heads/master"},"8a61fcdb5d13e9058233a1081796b5a07ec40a70":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-07-03 10:30:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/3"}}},"commit":{"parents":[{"commit":"191f359ad227ae1b3cc08e13b7f6e343f1b45af8","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/191f359ad227ae1b3cc08e13b7f6e343f1b45af8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 10:30:16.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8a61fcdb5d13e9058233a1081796b5a07ec40a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8a61fcdb5d13e9058233a1081796b5a07ec40a70"}]},"branch":"refs/heads/master"},"73fcd57f50e587f62954cd1b94ecfd6dea3fb0b9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-03 14:37:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/4"}}},"commit":{"parents":[{"commit":"1ac47e82131a1fdbe49da8b989fcfa0f4e7fcba1","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1ac47e82131a1fdbe49da8b989fcfa0f4e7fcba1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:37:54.000000000","tz":0},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/73fcd57f50e587f62954cd1b94ecfd6dea3fb0b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/73fcd57f50e587f62954cd1b94ecfd6dea3fb0b9"}]},"branch":"refs/heads/master","description":"Rebase"},"21bd5012f95238ecde0dc463f0f5cb06e1f56eb3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-03 14:40:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/5"}}},"commit":{"parents":[{"commit":"33a29f1dac55be6663ff4c7e55d827493b2d2596","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/33a29f1dac55be6663ff4c7e55d827493b2d2596"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:39:55.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/21bd5012f95238ecde0dc463f0f5cb06e1f56eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/21bd5012f95238ecde0dc463f0f5cb06e1f56eb3"}]},"branch":"refs/heads/master"},"b6d1569221f5bf37148fb19aebe7f1f038cc9155":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-07-04 09:13:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/6"}}},"commit":{"parents":[{"commit":"7a9adf1d6fc986120c29d9293929c36c7759fff2","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7a9adf1d6fc986120c29d9293929c36c7759fff2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 09:12:08.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b6d1569221f5bf37148fb19aebe7f1f038cc9155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b6d1569221f5bf37148fb19aebe7f1f038cc9155"}]},"branch":"refs/heads/master"},"389903d693aa9ee36037d92bb461580dad4c1d0f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-04 11:00:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/7"}}},"commit":{"parents":[{"commit":"acdfe9cc5c8ccc3ee9e5bd3ffd1850d748fc2cb1","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/acdfe9cc5c8ccc3ee9e5bd3ffd1850d748fc2cb1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 11:00:30.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/389903d693aa9ee36037d92bb461580dad4c1d0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/389903d693aa9ee36037d92bb461580dad4c1d0f"}]},"branch":"refs/heads/master"},"1021397854f3aea2720c8d6df732a99e87c7994b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-07-04 17:59:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/8"}}},"commit":{"parents":[{"commit":"d3809ebf6e6761352bfbf29f85d23ed0d573393e","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d3809ebf6e6761352bfbf29f85d23ed0d573393e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 17:59:02.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1021397854f3aea2720c8d6df732a99e87c7994b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1021397854f3aea2720c8d6df732a99e87c7994b"}]},"branch":"refs/heads/master"},"4f6e485a16adac01232928fbea1a607cbc3186d7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-07-04 20:35:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/9"}}},"commit":{"parents":[{"commit":"f7fe63ff3c64445dc51b653220b923be31deaa49","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f7fe63ff3c64445dc51b653220b923be31deaa49"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 20:28:47.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f6e485a16adac01232928fbea1a607cbc3186d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f6e485a16adac01232928fbea1a607cbc3186d7"}]},"branch":"refs/heads/master"},"cd786893e020d2b1cb8b1ac031b0ca1ce64f510d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-07-05 17:42:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/10"}}},"commit":{"parents":[{"commit":"552a39de727c97e9f27a3fff8de56e5509f34fae","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/552a39de727c97e9f27a3fff8de56e5509f34fae"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-05 17:42:04.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd786893e020d2b1cb8b1ac031b0ca1ce64f510d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd786893e020d2b1cb8b1ac031b0ca1ce64f510d"}]},"branch":"refs/heads/master"},"4324e428ea45e14e900f991afdc8add048cd5e11":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-07-07 08:30:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/11"}}},"commit":{"parents":[{"commit":"f2528bfbfe0af55ef8dd747836b24cff5db88b0e","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f2528bfbfe0af55ef8dd747836b24cff5db88b0e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-07 08:30:29.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4324e428ea45e14e900f991afdc8add048cd5e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4324e428ea45e14e900f991afdc8add048cd5e11"}]},"branch":"refs/heads/master"},"daa01fe92c47a6105820b0438acceabda0ba503e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-07-09 15:04:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/12"}}},"commit":{"parents":[{"commit":"df946d4019f129fbcd757c5c95f943886eb895e0","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/df946d4019f129fbcd757c5c95f943886eb895e0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 14:27:31.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/daa01fe92c47a6105820b0438acceabda0ba503e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/daa01fe92c47a6105820b0438acceabda0ba503e"}]},"branch":"refs/heads/master"},"0ee2dee3c36ab4af029c8bb4e515d127bdaccd15":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-07-09 17:03:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/13"}}},"commit":{"parents":[{"commit":"d52c1aaa765521997b3f0063d1e2bc94c3552a30","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d52c1aaa765521997b3f0063d1e2bc94c3552a30"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 17:03:38.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0ee2dee3c36ab4af029c8bb4e515d127bdaccd15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0ee2dee3c36ab4af029c8bb4e515d127bdaccd15"}]},"branch":"refs/heads/master"},"d01b82007a8d9412d0877f02b1c161852e51fb8d":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-07-10 10:37:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/14"}}},"commit":{"parents":[{"commit":"142a7db5036f743c1dddbab73bdbedf5778452d2","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/142a7db5036f743c1dddbab73bdbedf5778452d2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 10:35:29.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d01b82007a8d9412d0877f02b1c161852e51fb8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d01b82007a8d9412d0877f02b1c161852e51fb8d"}]},"branch":"refs/heads/master"},"7e489228f347978146aab3f4ce3e0505753e8196":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-07-10 12:46:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/15"}}},"commit":{"parents":[{"commit":"baeace702c518f516e94760a3c8e2b7658427ab4","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/baeace702c518f516e94760a3c8e2b7658427ab4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 12:41:23.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7e489228f347978146aab3f4ce3e0505753e8196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7e489228f347978146aab3f4ce3e0505753e8196"}]},"branch":"refs/heads/master"},"7c2c2a7487c22e018d4ddda669be243236d23340":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-07-17 09:32:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/16"}}},"commit":{"parents":[{"commit":"fe49db4cbdcad926a5ab2a0c08f3f639bc60ab45","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fe49db4cbdcad926a5ab2a0c08f3f639bc60ab45"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 09:32:05.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7c2c2a7487c22e018d4ddda669be243236d23340"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7c2c2a7487c22e018d4ddda669be243236d23340"}]},"branch":"refs/heads/master"},"8f1a6ec4254d005518b47edc62e2916c4e3d2a57":{"kind":"REWORK","_number":17,"created":"2025-07-17 10:13:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/949055/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/55/949055/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/55/949055/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/55/949055/17"}}},"commit":{"parents":[{"commit":"145fead384224de9fd73d0a475371c190642295d","subject":"Deprecate various \u0027[metadata]\u0027 options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/145fead384224de9fd73d0a475371c190642295d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-07 16:22:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 10:11:11.000000000","tz":60},"subject":"Deprecate various \u0027[backwards_compat]\u0027 options","message":"Deprecate various \u0027[backwards_compat]\u0027 options\n\nChange-Id: Icefe8027567cf7cc8153552712931d68174dd99b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8f1a6ec4254d005518b47edc62e2916c4e3d2a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8f1a6ec4254d005518b47edc62e2916c4e3d2a57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
