)]}'
{"id":"openstack%2Fpbr~949170","triplet_id":"openstack%2Fpbr~master~I5df54214b85858052fa60cc9caec889b81fab73b","project":"openstack/pbr","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-03 13:52:06.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I5df54214b85858052fa60cc9caec889b81fab73b","subject":"docs: Remove use of sphinxcontrib-apidoc","status":"MERGED","created":"2025-05-08 13:05:15.000000000","updated":"2025-07-03 14:37:33.000000000","submitted":"2025-07-03 14:36:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949170-pip-23.1-support","meta_rev_id":"0e0094d02272521fdcf7ed1ee04f5ab2b1eed224","_number":949170,"virtual_id_number":949170,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-03 14:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-07-03 10:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-07-03 10:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-07-03 13:52:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 13:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 14:47:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-07-03 09:46:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"f73dfa47accb37b04c6d5c1ab7abef9f7e8dfe18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-08 13:05:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33a82823246a17a03c0b2e5b5fa2e2b94e882e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 13:37:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afb6f52d00b847a2853ee4acb9a73f2b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86538b3f4fc647028035a9375b7bd6bc : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1d7c1d34004ec7aa7bd2b92bc33616 : SUCCESS in 1m 45s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e52a7bff971e461e8bad783a57773cc0 : FAILURE in 27m 07s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/448e015e4ef34dc2af84f2b49e484d57 : FAILURE in 29m 21s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/f9ffff30f193490ebb7e01df8c2e6be2 : FAILURE in 12m 14s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/28bbc237d6fe4ed9b938d422841b1ac6 : FAILURE in 22m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9328eea3489a695e25ed3597825f1f9fe8c624bd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-08 14:47:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17d3ece63196f9b599176b1f627e20ce6f6450f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 09:38:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a4056d0602390056175ce4e44cb61c406012fe8b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-07-03 09:46:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b80ae258a0a15d1e56ecea106efdf9e39ab622d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 09:59:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d581d2286ab4dd38f9bf64c897deecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb2f95d68a1c424da5c0f413368b4ded : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ad46aa4f6b41b5b085d9f328ca567e : SUCCESS in 1m 51s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/16656418d633421ca729e7cb584cdf14 : SUCCESS in 14m 55s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/a23c88d0e7684aa181cfe7540cf0ab9e : SUCCESS in 16m 30s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/f5a05a0080e14152b8d5394796174216 : SUCCESS in 18m 51s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4bd627adb6d6474592304e49a3702fff : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":2},{"id":"671c38a753677e3640becb4561c7605651efff00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 10:30:25.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a62387a9781c13a79d498df53a90179a97b76be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 11:14:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5388e9da7ad24907953e497b80b061a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4984054c74b45e68ca0f2e24ecf2de9 : SUCCESS in 2m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1378c8512f5847c8a10998d4031986ca : SUCCESS in 3m 41s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/5232a40795614124be29378df47d301e : SUCCESS in 26m 59s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4f96b7a397534d2f9e5548091eea644a : SUCCESS in 24m 03s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/ae531e5481b94e2b983d33227a394bc1 : SUCCESS in 36m 16s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/de730abe4ecf4c7ca4b72c7f8083f35c : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":3},{"id":"23dc1146e562fb739100aa3aef6ec814bf338332","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 13:52:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2147c04902128d2bf62860d6df30be637a7efeb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:53:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4bc16e1aa4171b19a682d21f271b0559a671a9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:36:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f8ace5a330f40c0af2a940e53b09069\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1a7fe534ed477bb653324d64add364 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc85c937734644f59b872d99dcc1c3a1 : SUCCESS in 1m 49s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/72c0c4363eb842f499978f5edf7a273a : SUCCESS in 14m 18s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/dca47bbc16bb448abe6e3834e4c99c30 : SUCCESS in 27m 31s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/fccc2c79fbbf42c1a40f9f47cc19774f : SUCCESS in 29m 32s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/d06afde383724482bd7e5c171b4dae1b : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1b6d4b712f144e507a66e6b91cb17b21caed2418","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:36:38.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"0e0094d02272521fdcf7ed1ee04f5ab2b1eed224","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:37:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e2151a8019244c7ae31ff656dbb84ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a4b56af0214b75a2d153f280cd3af2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0c1fb38317974b824bc0c22dcae569552b874c81","revisions":{"7aff914dbee0c594283fc4233bef93aa3fe34e89":{"kind":"REWORK","_number":1,"created":"2025-05-08 13:05:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/949170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/70/949170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/70/949170/1"}}},"commit":{"parents":[{"commit":"5d4a1815afa920cf20e889be20617105446f7ce2","subject":"docs: Drop SetupTools from pyproject.toml example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d4a1815afa920cf20e889be20617105446f7ce2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 12:59:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:03:42.000000000","tz":60},"subject":"docs: Remove use of sphinxcontrib-apidoc","message":"docs: Remove use of sphinxcontrib-apidoc\n\nIn favour of the new upstream extension.\n\nChange-Id: I5df54214b85858052fa60cc9caec889b81fab73b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7aff914dbee0c594283fc4233bef93aa3fe34e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7aff914dbee0c594283fc4233bef93aa3fe34e89"}]},"branch":"refs/heads/master"},"717bf51bb6c04ba7abdc79e903392b56a648610f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-03 09:38:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/949170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/70/949170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/70/949170/2"}}},"commit":{"parents":[{"commit":"0b3da64bfbbd59da72d9eeb82c692fa42e9a075f","subject":"Merge \"Don\u0027t pin setuptools for all tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0b3da64bfbbd59da72d9eeb82c692fa42e9a075f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 12:59:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 09:38:23.000000000","tz":60},"subject":"docs: Remove use of sphinxcontrib-apidoc","message":"docs: Remove use of sphinxcontrib-apidoc\n\nIn favour of the new upstream extension.\n\nChange-Id: I5df54214b85858052fa60cc9caec889b81fab73b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/717bf51bb6c04ba7abdc79e903392b56a648610f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/717bf51bb6c04ba7abdc79e903392b56a648610f"}]},"branch":"refs/heads/master"},"0c1fb38317974b824bc0c22dcae569552b874c81":{"kind":"NO_CHANGE","_number":3,"created":"2025-07-03 10:30:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/949170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/70/949170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/949170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/70/949170/3"}}},"commit":{"parents":[{"commit":"0b3da64bfbbd59da72d9eeb82c692fa42e9a075f","subject":"Merge \"Don\u0027t pin setuptools for all tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0b3da64bfbbd59da72d9eeb82c692fa42e9a075f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 12:59:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 10:14:28.000000000","tz":60},"subject":"docs: Remove use of sphinxcontrib-apidoc","message":"docs: Remove use of sphinxcontrib-apidoc\n\nIn favour of the new upstream extension.\n\nChange-Id: I5df54214b85858052fa60cc9caec889b81fab73b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0c1fb38317974b824bc0c22dcae569552b874c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0c1fb38317974b824bc0c22dcae569552b874c81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
